)]}'
{"id":"openstack%2Fbifrost~787505","triplet_id":"openstack%2Fbifrost~master~I3cba11ac02a09d142637e2169a7b901a5b41bd59","project":"openstack/bifrost","branch":"master","hashtags":[],"change_id":"I3cba11ac02a09d142637e2169a7b901a5b41bd59","subject":"Install squashfs-tools for Ubuntu image builds","status":"MERGED","created":"2021-04-22 08:38:57.000000000","updated":"2021-04-26 10:07:16.000000000","submitted":"2021-04-26 10:05:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787505","meta_rev_id":"933949e5ebb61e56fecb7b64e2a34a772c934bc7","_number":787505,"virtual_id_number":787505,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 10:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-04-26 07:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-26 08:57:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-26 08:57:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 09:52:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-23 08:15:02.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-04-26 08:57:18.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"d004854dd7a76af0405d90e4dca77114d0d5dd3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-22 08:38:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2924fba69c36322ff65ea0e56cdeb73c0f23858c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:52:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84cb0302a9314452aba072f0f8114510 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b850db96b1fe4be3ba1e0ddff53d9a8c : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6fd6181b9c84c7ea7b6e6da629a2251 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/605c3ed8cb88420c8f7f4ad0c3788778 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb772a1489214aada56fb178a659d4cb : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21afb821f9a545dba5fdf49c3867edc1 : SUCCESS in 7m 51s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/46d38154117e4e9aa91e1f40e4ab0cc0 : SUCCESS in 6m 14s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/757f3726f78f4242a2d7e262e9eae6e2 : SUCCESS in 31m 44s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/e5e6814c7e334b03a47edc7e1ba590f9 : SUCCESS in 30m 25s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae43fe3950f147d1bb6baa04a07383e9 : SUCCESS in 25m 45s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/cecf0b1a604a4fef82c62bf6ab8c059f : SUCCESS in 35m 50s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc043ecd76f145e69e293a179bf33542 : SUCCESS in 28m 14s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/3416a4e2ebd447cd8e97175ee7cb66b5 : SUCCESS in 30m 00s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05de47929e2d48c389b62a5cbc7a116b : SUCCESS in 44m 35s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/a31fdef4795d4bb292ce03a30f0917b6 : SUCCESS in 1h 03m 25s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9a4ae866cd7494699d34b91bcaf4048 : SUCCESS in 27m 26s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/dfe3582ab20444a5997c1a7924f7a36e : FAILURE in 13m 43s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/8dbd6aab7533440eb87d49feee395bfc : SUCCESS in 27m 44s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/ba12d842403d4554bed0faad27fc3652 : SUCCESS in 24m 49s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/c924541312eb4a158c9578a755578ca4 : SUCCESS in 27m 05s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8b766c99f3c4fb98684af0f1b19b0c1 : SUCCESS in 37m 25s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/59d994ac444b422a97bcc998ebe48167 : SUCCESS in 33m 09s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb86a561fdcc492b8b117afa11bfcecf : SUCCESS in 40m 03s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/5e998a00beab45c1914e9a93e38406d2 : SUCCESS in 33m 39s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46a9d283cabf451b81dc31dd44ad1316 : SUCCESS in 27m 53s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/97f88da00891414aa94b06bc522735c5 : SUCCESS in 33m 33s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/4ad5b4f74c614398b607aa4e439f8790 : SUCCESS in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76d5de7f4be31a7057225888b17ea34597a26cb4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-04-23 08:15:02.000000000","message":"Patch Set 1: Code-Review-1\n\nthis breaks suse though as there\u0027s no squashfs-tools package\nmaybe differentiate by distribution?","accounts_in_message":[],"_revision_number":1},{"id":"44cd58e05821b1bd16435409eae7096511a8d688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-23 15:32:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97018a52e961c3fe671b458a7f72a6c321d8ae66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 18:47:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/823f55e796e14d1fb64c30620515f29a : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05e06670d04c4b37818091c919bf61f0 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db4b4062f1664162aa0ad34ddb6be5e8 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7856212e5f1b42368cc11697f140a862 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7718f7426e5434b8edf7baf6403c9d1 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/275952618b5045d89f843197edb20501 : SUCCESS in 4m 36s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/f27b11c2bf2d479b9fb4f2face7b778d : SUCCESS in 6m 14s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc25ee2ca06f4229ad5a32e6bd68905f : SUCCESS in 30m 30s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/aab06d950402401b94d7f822bce5a2f8 : SUCCESS in 29m 56s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a66040758b9e4b5aa9a68ff98e86e3f1 : SUCCESS in 39m 22s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/5f2c14a049f043f0acbcafeb1fc5a7ef : SUCCESS in 31m 07s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a180d16ae24c45aeb11b16a09114056a : SUCCESS in 27m 22s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/d19bc90d35604d789b75d90da113a20e : SUCCESS in 27m 36s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/542338c19fc043728f1a3b4828a32f53 : SUCCESS in 49m 08s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/d77a436876ab4c2b9b4b72060a321c75 : SUCCESS in 1h 01m 30s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0f9fc7feca64ba1af4d4667eeccbe10 : SUCCESS in 27m 24s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/a7ad9d5c514241569874f4cbacb489a5 : SUCCESS in 27m 53s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/47b942328e254ec48623550b45ed09a3 : SUCCESS in 28m 38s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/66ed293e2442400dbaf06f75351bdc71 : SUCCESS in 25m 24s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/38fa1ec19d9948e7ab9ff8ea7ab181d3 : SUCCESS in 28m 14s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/967d47598ae1440598f7e8ec2f95d26d : SUCCESS in 31m 15s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/36a3b3996cce4e89b8cc1d9431078d8a : SUCCESS in 33m 04s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2ed60199af74a5381f02041be6f560d : SUCCESS in 31m 19s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/3140a29f79df4d10b3fe08fd65363fcf : SUCCESS in 33m 46s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c75fae96b68b48f6993b2b54b541cb01 : SUCCESS in 26m 36s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/8d8c450e60f94fbfa755817baa6155c4 : SUCCESS in 27m 39s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/2fad84557ec34f10adbabac464e59e9d : SUCCESS in 27m 14s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/20bda3b128f641748583d20061e2eec1 : FAILURE in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16da5cabe481da029a8c4e3f2ba1e7ae246f3e55","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-04-26 07:33:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4416ea0bc9cd6f05b44d9e03cece3a0410aa061f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-04-26 08:57:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bb3a0d064c46ad9e5765a0e667ad0003969f4f10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 09:00:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef884a0ad0e998652e4203b470b8a2305c75bd44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:05:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec8c3543742c443f8189844c9a683f57 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b61fbdb358ca40daa5ee5ddb90bbd09b : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a649359cb050459ca09a71f6fcaeca5c : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60aff1da9ccf4786b8812cb34db52596 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44f74c12110f43f0bc769a2641c3a4a2 : SUCCESS in 5m 27s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/390947c124f040b5a69ceb9de76d94cc : SUCCESS in 6m 13s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb086fe0853f4b61b721a8ac2a8daad3 : SUCCESS in 27m 23s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/2be9d1b1c7bd4c39b8b40338a2332c24 : SUCCESS in 30m 02s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a20d68d85c6b47139e7e105b49a1a6ac : SUCCESS in 31m 33s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/e21fcf98d2ef43c59c35dc7fc18f033b : SUCCESS in 30m 52s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa52032a53f743dfa28f88d304f7f9c7 : SUCCESS in 49m 06s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/2416511153bb452ea393ea9ec49aa979 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":2},{"id":"73d36d0a219404c6a017d55d98f87f381648d8df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:05:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"933949e5ebb61e56fecb7b64e2a34a772c934bc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:07:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c9d6af1cf7d4899bcc87495ba14f176 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0b175454c7648edb3905b055f5ed533 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b166a0629bafdc0b18de3835e9210dfeaf811322","revisions":{"9094eb22a10c060fc5f0e6b4813969ae68196503":{"kind":"REWORK","_number":1,"created":"2021-04-22 08:38:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/787505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/05/787505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/787505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/787505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/787505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/05/787505/1"}}},"commit":{"parents":[{"commit":"95c429b5ba18719ca05eb920437e41a58a23076e","subject":"Use explicit path for which","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/95c429b5ba18719ca05eb920437e41a58a23076e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-22 08:32:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-22 08:38:27.000000000","tz":60},"subject":"Install squashfs-tools for Ubuntu image builds","message":"Install squashfs-tools for Ubuntu image builds\n\nUbuntu cloud images are distributed in a squashfs file system. Using the\nubuntu DIB element requires unpacking the image, which requires the\nsquashfs-tools package. This change ensures it is installed.\n\nStory: 2008838\nTask: 42345\n\nChange-Id: I3cba11ac02a09d142637e2169a7b901a5b41bd59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9094eb22a10c060fc5f0e6b4813969ae68196503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9094eb22a10c060fc5f0e6b4813969ae68196503"}]},"branch":"refs/heads/master"},"b166a0629bafdc0b18de3835e9210dfeaf811322":{"kind":"REWORK","_number":2,"created":"2021-04-23 15:32:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/787505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/05/787505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/787505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/787505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/787505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/05/787505/2"}}},"commit":{"parents":[{"commit":"95c429b5ba18719ca05eb920437e41a58a23076e","subject":"Use explicit path for which","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/95c429b5ba18719ca05eb920437e41a58a23076e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-22 08:32:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 15:32:25.000000000","tz":60},"subject":"Install squashfs-tools for Ubuntu image builds","message":"Install squashfs-tools for Ubuntu image builds\n\nUbuntu cloud images are distributed in a squashfs file system. Using the\nubuntu DIB element requires unpacking the image, which requires the\nsquashfs-tools package. This change ensures it is installed.\n\nStory: 2008838\nTask: 42345\n\nChange-Id: I3cba11ac02a09d142637e2169a7b901a5b41bd59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b166a0629bafdc0b18de3835e9210dfeaf811322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b166a0629bafdc0b18de3835e9210dfeaf811322"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
