)]}'
{"id":"openstack%2Fbifrost~787509","triplet_id":"openstack%2Fbifrost~master~Ib6c9165ec61170da879240b189fb5b2be5e46b75","project":"openstack/bifrost","branch":"master","topic":"remove-bionic-integration","hashtags":[],"change_id":"Ib6c9165ec61170da879240b189fb5b2be5e46b75","subject":"Remove bionic integration job","status":"ABANDONED","created":"2021-04-22 09:00:19.000000000","updated":"2022-05-18 07:41:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d45817fc4d6dd862fc68007ba79a697e4199fb98","_number":787509,"virtual_id_number":787509,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-22 10:17:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-04-22 09:05:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":-1,"date":"2021-04-28 08:36:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 09:05:35.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-04-22 10:17:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41b8abdbf46245bfbc8d6a89c91fc0f203bf83c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-04-22 09:00:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dead0d46b9a97eb285dfd0136c1883446fdbc98c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-04-22 09:05:35.000000000","message":"Patch Set 1: Code-Review-1\n\nDo we really need to stop supporting Bionic? If so, please update the documentation and provide a release note.","accounts_in_message":[],"_revision_number":1},{"id":"c49dcf99dfda6c8c4e171f922e836929823949f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:17:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b8bd08b1ff949e29639e48922fb1a4b : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c26e32e6799452d8677e04d0eb55440 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e179402e1dc24832ba34485c60b72aa8 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a474917e4f44ee1bbc56e2c4f618bff : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25a1a0ce4814163bf9d1454aa631843 : SUCCESS in 5m 33s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/16444b3194eb4468955b3692d64d87a8 : SUCCESS in 4m 59s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8db3416c38d04e319f956ec46f4fb097 : SUCCESS in 29m 56s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/c23cad0e6a5940dca234c3cc876f1c9a : SUCCESS in 28m 00s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26171fac9a0548ce8451bfcc7d55390d : SUCCESS in 27m 31s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/f4ee8dbe4dd24fbf97d79a47fdc39dfe : SUCCESS in 34m 58s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1a095aae0d0486899c277ae972c4963 : SUCCESS in 28m 20s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/b3c09fb9b2144d168ffdb7ad1cbcbaba : SUCCESS in 28m 45s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/003ead07203f4a8490a18c8fc92aa580 : SUCCESS in 52m 56s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/d12664bd42714c01bb775518275ee5a3 : SUCCESS in 1h 04m 29s\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/775fbfcdd0984f67be04598566cb0d9d : SUCCESS in 37m 17s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/b51ba61061d94ddca397f3b04cf7b49f : SUCCESS in 30m 25s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/9651472e2c5b4f62897bd64a6167d455 : SUCCESS in 26m 07s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/926ba1ae4a404db2b69af1cde648d929 : SUCCESS in 29m 32s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c170236f9f174bfea1e92f500ab37cd2 : SUCCESS in 31m 11s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/15d4dc6b87e24f10af5e6fe1ecea0d07 : SUCCESS in 32m 53s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d93183cb1941492993613f14b2bbc150 : SUCCESS in 32m 32s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/cdd3b46920834908b425860efc2a94ba : SUCCESS in 35m 28s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/241da9d6a6ee446aab457089a6c25978 : SUCCESS in 31m 55s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/1b73a2416da04576bf895f0583c4f5c1 : SUCCESS in 27m 46s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/b8a7643f567c43f3b9ac1715686b7748 : SUCCESS in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"891a0f3d5344b1d26ec5940df2e6619750f912b2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-04-28 08:36:35.000000000","message":"Patch Set 1: Workflow-1\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Do we really need to stop supporting Bionic? If so, please update the documentation and provide a release note.\n\nehm... in some way I was able to confuse the end of support between xenial and bionic, so no, we don\u0027t need to do this now\ngoing to put this on standby","accounts_in_message":[],"_revision_number":1},{"id":"d45817fc4d6dd862fc68007ba79a697e4199fb98","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-05-18 07:41:20.000000000","message":"Abandoned\n\nIncluded into https://review.opendev.org/c/openstack/bifrost/+/842272","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1417edbb8e3730ec08260dd810e95833793f697","revisions":{"f1417edbb8e3730ec08260dd810e95833793f697":{"kind":"REWORK","_number":1,"created":"2021-04-22 09:00:19.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/09/787509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/09/787509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/787509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/787509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/787509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/09/787509/1"}}},"commit":{"parents":[{"commit":"95c429b5ba18719ca05eb920437e41a58a23076e","subject":"Use explicit path for which","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/95c429b5ba18719ca05eb920437e41a58a23076e"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-04-22 08:57:53.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-04-22 08:57:53.000000000","tz":120},"subject":"Remove bionic integration job","message":"Remove bionic integration job\n\nIt\u0027s non-voting since a while and no other project is using it.\n\nChange-Id: Ib6c9165ec61170da879240b189fb5b2be5e46b75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f1417edbb8e3730ec08260dd810e95833793f697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f1417edbb8e3730ec08260dd810e95833793f697"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
