)]}'
{"id":"openstack%2Fbifrost~787854","triplet_id":"openstack%2Fbifrost~stable%2Fvictoria~I3cba11ac02a09d142637e2169a7b901a5b41bd59","project":"openstack/bifrost","branch":"stable/victoria","hashtags":[],"change_id":"I3cba11ac02a09d142637e2169a7b901a5b41bd59","subject":"Install squashfs-tools for Ubuntu image builds","status":"MERGED","created":"2021-04-26 10:08:10.000000000","updated":"2021-04-26 11:25:44.000000000","submitted":"2021-04-26 11:24:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787854","cherry_pick_of_change":787505,"cherry_pick_of_patch_set":2,"meta_rev_id":"447cd0de0457a5793a017c23d4fda93afb3b691e","_number":787854,"virtual_id_number":787854,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 11:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-04-26 10:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2021-04-26 10:16:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 10:08:10.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-04-26 10:16:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-04-26 10:49:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ace0b8722afb0a8418464ddfb7d1b71b4f43e3bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-04-26 10:08:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"addeb5d59b7172593e78747b466d5826fb0ae68a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-04-26 10:16:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"edb2af894a81cc4afd2b8fcbf87960dbddb58714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:49:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2e7df05be04c27b2659e775091d036 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a42ff0783364f47a4a411bd4256f3d7 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/314c2278a7e443e9b074eb70cdf14a0f : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d49a3e57bf6849578b72888719e28e34 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6959d93c0fad425cb52c6ff722ed4cbf : SUCCESS in 5m 02s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/9dc6049bc43b469da479ae4035d7fb17 : SUCCESS in 5m 59s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f028d31ea64e454c8cdf9653870a251c : SUCCESS in 28m 57s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/b49422cfb9af4354a8697abd2b573838 : SUCCESS in 28m 11s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85c953ef2b2d4350b37f988c17cf3637 : SUCCESS in 29m 27s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/62578ca5e20f4d609434ec2fbc165fbe : SUCCESS in 29m 22s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b197f46815949a78e52c11fd785507b : SUCCESS in 24m 48s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/24ac311cd31d44bf9419cc812cd1c8c1 : FAILURE in 13m 04s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/851f3aa6689d4e5a919bbe417d5fbcd3 : SUCCESS in 28m 01s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/3a5dc31912dc435291640a6d08a03579 : SUCCESS in 23m 35s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28e642c55a994ed78147a0727d03c017 : SUCCESS in 31m 18s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/c57951fd2bfb469a8b4f4856bf2cb397 : SUCCESS in 31m 00s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3318e2a9187445bb26f55a26eac7ffa : SUCCESS in 28m 59s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/af782c7f13354b4cb9ddd376c4f149fd : SUCCESS in 30m 15s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94f880e656fb4b3f83200b1d6e789357 : SUCCESS in 24m 36s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/96bdc2b746324a6c83e1a39e3070f8dc : SUCCESS in 24m 26s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/94e58e20ecd349ccbb997124fc6d957f : SUCCESS in 25m 31s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/8e1601bf1cb246f2840cb777242d990e : SUCCESS in 27m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0fa1448f669c5b3e500c3af1ba34592972057ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:49:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b644cd03458e96dfe114e84ed431ac3fa332b3e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 11:24:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926fe47f2a28497dafcd7622ddb5fb6d : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70394d5568c04a3eb3b0752f66328f38 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf8dd07482ab48ccbecce01de1817f56 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/494d9ffa8c144e4abe569115509797da : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13e2e87da94a40b6b092868ddcf83c69 : SUCCESS in 3m 55s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/e419261059a54f518fe3a42bcc851b6b : SUCCESS in 5m 13s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f66d5f73fe30440b99ef2894c3bddaa3 : SUCCESS in 21m 40s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/67c525e1fdc447fc87d50ea609b364a1 : SUCCESS in 31m 32s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/802c174bc63e417e83ceca9f3bc9ac34 : SUCCESS in 27m 05s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/03e0283625fc4c6da84008b2b167ebf0 : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":1},{"id":"01fe9525dff3ac975a1643d62fba73b664b63f69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 11:24:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"447cd0de0457a5793a017c23d4fda93afb3b691e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 11:25:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39809243e68343c683d0e7c1949552d4 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6556735d4b5e49a7b6653211f4e70c0a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2b16602363c7c48fcc50bd83bacd48561f74550","revisions":{"a2b16602363c7c48fcc50bd83bacd48561f74550":{"kind":"REWORK","_number":1,"created":"2021-04-26 10:08:10.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/54/787854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/787854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/787854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/787854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/787854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/787854/1"}}},"commit":{"parents":[{"commit":"c2a4f845befe36e2b76645060f7c71d1e62902a5","subject":"Use explicit path for which","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c2a4f845befe36e2b76645060f7c71d1e62902a5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-22 08:32:48.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-04-26 10:08:10.000000000","tz":0},"subject":"Install squashfs-tools for Ubuntu image builds","message":"Install squashfs-tools for Ubuntu image builds\n\nUbuntu cloud images are distributed in a squashfs file system. Using the\nubuntu DIB element requires unpacking the image, which requires the\nsquashfs-tools package. This change ensures it is installed.\n\nStory: 2008838\nTask: 42345\n\nChange-Id: I3cba11ac02a09d142637e2169a7b901a5b41bd59\n(cherry picked from commit b166a0629bafdc0b18de3835e9210dfeaf811322)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a2b16602363c7c48fcc50bd83bacd48561f74550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a2b16602363c7c48fcc50bd83bacd48561f74550"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
