)]}'
{"id":"openstack%2Fbifrost~799975","triplet_id":"openstack%2Fbifrost~master~Ib8e1b002afc053617aa1f39c342bd1809d0154a4","project":"openstack/bifrost","branch":"master","topic":"collect-pkgs","hashtags":["ironic-week-prio"],"change_id":"Ib8e1b002afc053617aa1f39c342bd1809d0154a4","subject":"Collect explicit list of installed packages","status":"MERGED","created":"2021-07-08 09:15:32.000000000","updated":"2021-07-21 08:56:36.000000000","submitted":"2021-07-21 08:54:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799975-collect-pkgs","meta_rev_id":"6f522b83036c9661caf500afab1b7252f2378fbe","_number":799975,"virtual_id_number":799975,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-21 08:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-11 09:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":2,"date":"2021-07-21 07:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":1,"date":"2021-07-21 07:54:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-08 09:56:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-11 09:13:56.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-07-21 07:54:47.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"599944446b8b2e3efef25cc96be1ec3d6c7ef272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-07-08 09:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af13954ea41e98e17eb24e83def99bcc021e9a0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 09:56:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/147203ca94804cc98310b7751581bac3 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4307d44effb44c488ab04d8ee076ee0 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76374caff30c43b0b5069ee27cdcdc88 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/898e16d62bb749f892dac6d79b967e93 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de3b224c8254800b40a4351defc4104 : SUCCESS in 6m 48s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/ceba502d32c14b49abf7ea50ca828a52 : SUCCESS in 5m 47s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34031e3f736c45069f9e253f72d16b90 : SUCCESS in 25m 35s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/e33e05f7594544f7aae8a7ac647a3e24 : FAILURE in 7m 50s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f70e04fdf1cf4724850f0796981e019d : SUCCESS in 23m 44s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/f94b261d3408440aabe9936d85c935db : FAILURE in 8m 04s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cff32257d3274fb89207af8d410da2df : SUCCESS in 20m 48s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/d1159747f3d9417fb11efefbf7821499 : FAILURE in 7m 58s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2962376329414b42abe3461bfa2a868c : SUCCESS in 39m 16s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/df98536ef09443038d82a501afa032eb : FAILURE in 29m 10s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd1a65b7513d4402a59b192af8399ecd : SUCCESS in 23m 40s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/01a6a243772e4db78b9c14529349cda7 : RETRY_LIMIT in 2m 47s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/5056898b97544bce99ad24dd582a0a27 : SUCCESS in 26m 40s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/fd1fb5a05a7b4293ac2ede2ebd2296ae : SUCCESS in 25m 30s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/f635241e1f0a4ccdaf50c542f0b775dc : SUCCESS in 23m 39s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d50abfa156145a68a524d29f5807005 : SUCCESS in 28m 41s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/e11e0696116542788a340f484f1f9e37 : FAILURE in 7m 50s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ae73de9a74f4e50aa174cb4c6596175 : SUCCESS in 28m 58s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/d98865be77e6440c8f8be9ef666a3fae : FAILURE in 8m 17s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60307cec194f495985c7d2bca51b0a7f : SUCCESS in 24m 52s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/904dd3b40972409b8e1d1d02a4512b68 : SUCCESS in 22m 09s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/5b2f9ef920454340989b10d791089205 : SUCCESS in 28m 54s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/7776759270864aa0bfb6ea614abdf963 : FAILURE in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d1f97336348fcff0a81e09490def6cd9711addb","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-07-11 09:13:56.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62e00ccd60dcff211da8f94597917036cf988622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-11 10:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe8fa2300f34bf3b55c6cf666afc820 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/698b0aa935f746f0a5dacb146ee4df90 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8caa1ddf17bd43fdae00a5d52d6a0d25 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f092953e5746fab296707ea5412e7f : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68fadd97c643440abb57ba721a40edd3 : SUCCESS in 5m 16s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/e55166e9e7ab4d01820945ee1182644d : SUCCESS in 6m 14s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/416e4b3a218f4120a61365ae167a984c : SUCCESS in 29m 33s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/583cf1c580fa45c3b5baad69f608a39f : SUCCESS in 27m 50s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a38fcbd1211486ba60947e4e6472cda : SUCCESS in 21m 08s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/837c41baa77943f38842a042d972ce8c : SUCCESS in 31m 16s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/309805e6185f4a75a9cf995657856e69 : SUCCESS in 20m 07s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/45fa99d11378458f80f853e96a53f84b : SUCCESS in 20m 40s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca8715cd0d0a41e887e822c5827b9283 : SUCCESS in 52m 01s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/b75ce3241916436898b76a1fd0c68051 : SUCCESS in 49m 20s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a388e1f196c742d1a8f21f6f08f04bb5 : SUCCESS in 27m 13s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/155f48be70be4c10b1565d52f5ffa8c1 : RETRY_LIMIT in 3m 07s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/edbdf992f632440ea815fbfeff1f6268 : SUCCESS in 24m 44s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/215381ef0798492cb2806daec9ed80f1 : SUCCESS in 25m 52s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/0d2549d7e0d1404c88d523b9223312c6 : SUCCESS in 28m 19s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/039a1162c0604bdeb785218f58baa56d : SUCCESS in 26m 56s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/08911a480578497c8cac4c4138e66a18 : SUCCESS in 32m 32s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ea733cb8a954f4e848050f91accd904 : SUCCESS in 31m 59s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/11a32580d4a84d61b32bc0ebc1640733 : SUCCESS in 28m 57s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18917a655c324af2b4c322938f29ab66 : SUCCESS in 26m 55s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/acb4479a897541a9995a25b03aa7d5ce : SUCCESS in 24m 04s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/73f18b9d382446efbc1766e59c38ff16 : SUCCESS in 28m 40s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/f18d504e71864c379d103e43051a8cd2 : SUCCESS in 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7309fc1ee4de81f44e7be542fc0a3b8f9c8c4c8","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-07-12 15:27:49.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"316fa1f3d2439dc9f6d47a74eaff07131ecde23c","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-07-21 07:54:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"16c29863e9d9a3931d75815cea1b8fe486d10cc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 07:55:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"924e113ffb30a2efe55ddadb44b604daf47c2799","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 08:54:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8790d7da61c4dc4bbe6a3389d5f61e2 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6b1f3caa2bb45d192c3a9cd2b1772d6 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/774ba437159b483c87c2d34932c51767 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5978f563749c443fa71c1fe20a0d4611 : SUCCESS in 6m 09s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/76783c9c804b4b5589b35af91fbba5ec : SUCCESS in 5m 41s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47bac21eb34f40dea24f78ddd8f0c916 : SUCCESS in 22m 50s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/4b8f6ec7f6144066b3336e9e237f7b6f : SUCCESS in 26m 17s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c1d474044b4a33ac44f23836161784 : SUCCESS in 24m 54s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/5ec7e80ef96a47dbab6c60a0fae3f1a5 : SUCCESS in 27m 51s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/783f7879bf5d4c5e92ebc4dfa61044ae : SUCCESS in 42m 25s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/12ca57fc7fde4782bcc55d1ee4e320d3 : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":1},{"id":"075a00b004d4e0bf63cd02b8ea0a999c84e6cbd8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 08:54:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6f522b83036c9661caf500afab1b7252f2378fbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 08:56:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ad26985b5bf4dd3814ade8db3cb6823 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d751208acae110d68c4b60db0ea3443ab41d5c08","revisions":{"d751208acae110d68c4b60db0ea3443ab41d5c08":{"kind":"REWORK","_number":1,"created":"2021-07-08 09:15:32.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/75/799975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/75/799975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/75/799975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/75/799975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/75/799975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/75/799975/1"}}},"commit":{"parents":[{"commit":"a14d96a2dede1fcdfc5e42f39372a6e4f0dad86e","subject":"Merge \"Skip running jobs on ironic and inspector unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a14d96a2dede1fcdfc5e42f39372a6e4f0dad86e"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-07-08 09:15:09.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-07-08 09:15:09.000000000","tz":120},"subject":"Collect explicit list of installed packages","message":"Collect explicit list of installed packages\n\nChange-Id: Ib8e1b002afc053617aa1f39c342bd1809d0154a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d751208acae110d68c4b60db0ea3443ab41d5c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d751208acae110d68c4b60db0ea3443ab41d5c08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
