)]}'
{"id":"openstack%2Fbifrost~802754","triplet_id":"openstack%2Fbifrost~master~Ib8e55dc9909381740d79fc660e46ecd329b89283","project":"openstack/bifrost","branch":"master","topic":"sushy-state-dir","hashtags":[],"change_id":"Ib8e55dc9909381740d79fc660e46ecd329b89283","subject":"Keep sushy-emulator state directory in /var/lib","status":"MERGED","created":"2021-07-28 16:37:09.000000000","updated":"2021-08-02 13:25:23.000000000","submitted":"2021-08-02 13:23:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802754-sushy-state-dir","meta_rev_id":"4727a1e8f44135e83616442b0b6b7d6b6a40b356","_number":802754,"virtual_id_number":802754,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 13:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-08-02 12:17:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-02 11:58:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2021-08-02 12:17:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 17:49:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-02 11:58:04.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-08-02 12:17:04.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"b8ac6bd26b746d7bc26f9700d3f4c6c2be64474d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-07-28 16:37:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05790b0381edc7b41be05249efa279ac80284965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-07-28 16:44:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfe6c322788d3ea87effd46c7ab9e29cdbdb1654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 17:49:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eecc31bbaae14cc19c34f01843c71168 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5294040f9f254adda1918217dde3d56e : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a31f082969e2436a93bfd1b7e6ba8d0a : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c91deafc55e4447ca9c5c1ae3d44ec7c : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9954a5b47e427da45bdbe2fa5122f9 : SUCCESS in 5m 58s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/f58942a6e6e0495283be7616bd946f42 : SUCCESS in 5m 57s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9aeff536dfe4d868d47d19f6ee14389 : SUCCESS in 27m 48s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/09997cad3bbb492ca9caec582354c3f0 : SUCCESS in 30m 02s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/106b51df399746a2a2e38845bb2e09c3 : SUCCESS in 32m 41s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/9f331ae2362f4dde98381155e0649b04 : SUCCESS in 33m 30s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cf4d304430542c48913a2d58e4f657b : SUCCESS in 23m 37s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/d5b37fc0837e4810a20883ac7c43d1b5 : SUCCESS in 34m 18s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e24868653dde4e9f83e7533894a0a9e0 : SUCCESS in 42m 37s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/63a20f33d0bc46cb890dba1921ced48f : SUCCESS in 1h 01m 49s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d8e99cd44e944eb9548e7aa91a4a7a2 : SUCCESS in 28m 14s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/a07ee0105d69440ab830f6613dab5b07 : SUCCESS in 32m 14s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/760ce0f726624985bd1ccda1dfd67f7c : FAILURE in 13m 43s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/7e6432fbbc4144968b01488e86c0f486 : SUCCESS in 27m 13s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/75a297ad3817410ba4e3397ffab919d1 : FAILURE in 11m 57s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3344d08b7e244d19ab6bd58b277c2ec : SUCCESS in 34m 07s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/04219e39dd3946e182858439ee065d6c : SUCCESS in 32m 25s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e09c2d5d039c4a5aa65b157e6c3cef2b : SUCCESS in 30m 48s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/e17a89e6b7b84e9598e992ac6d3eacc8 : SUCCESS in 35m 12s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c75ee7a7f584421db3950423c363a000 : SUCCESS in 26m 12s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/b760c993c184450fb03d178ce8e68768 : SUCCESS in 28m 56s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/6a6f6c874d994003835200f6a95082e8 : SUCCESS in 27m 15s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/9d0c633f67064b4f96fc00e188370cf6 : SUCCESS in 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7083717f616fda1c65efaa06a3b02adad827b284","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-08-02 11:58:04.000000000","message":"Patch Set 2: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"41755b9ee63e03631d85e5633314bc812dcb3acc","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-08-02 12:17:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9c958c66abb999c29ceeb2cb111adb12fd347183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:17:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc3c928e36ec81bda188304749074a9476294ea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 13:23:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd5b4d6b6ea46698d36e68335bca9ad : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32b0e8370c704202ab488dff94a5f646 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6835133326b44e44a2ff69ec42d6f35f : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df740304f180456798e722e21e535d71 : SUCCESS in 6m 20s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/8ad66f01c63e43bd9cd278bc2ff79564 : SUCCESS in 5m 49s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b43ea9241aa64e85a0268003e8d0d9c4 : SUCCESS in 27m 28s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/b2ae124494d3401d9c8bd59b982e167a : SUCCESS in 33m 20s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d66e1b7cc1824929a553752aae186b23 : SUCCESS in 28m 18s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/ebcc5fd06a254749b513785470b5c271 : SUCCESS in 29m 00s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aed4b7e4021407faf85c920942a9eb6 : SUCCESS in 50m 52s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/c21c3b5d12aa4542814e448df41e4de0 : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":2},{"id":"c19cb3831ebbade866a957e57450bc5848c0957a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 13:23:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4727a1e8f44135e83616442b0b6b7d6b6a40b356","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 13:25:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e3270de343e47f79d59c24c1404c161 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"369693328544bf240642e143aab97590f07449c4","revisions":{"d053f58b6d4b5464b2c405348252d45bd9e0f9a5":{"kind":"REWORK","_number":1,"created":"2021-07-28 16:37:09.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/54/802754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/802754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/802754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/802754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/802754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/802754/1"}}},"commit":{"parents":[{"commit":"dfa9eddb3f4b2cbed0e4fe4ec1604aac70d83bec","subject":"Show debug messages during upgrade ci job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/dfa9eddb3f4b2cbed0e4fe4ec1604aac70d83bec"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-07-28 16:36:21.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-07-28 16:36:23.000000000","tz":120},"subject":"Keep sushy-emulator state directory in /var/lib","message":"Keep sushy-emulator state directory in /var/lib\n\nBy default it\u0027s in /tmp, which means that it can be removed on reboot\nor even during normal operation (after some time).\n\nChange-Id: Ib8e55dc9909381740d79fc660e46ecd329b89283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d053f58b6d4b5464b2c405348252d45bd9e0f9a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d053f58b6d4b5464b2c405348252d45bd9e0f9a5"}]},"branch":"refs/heads/master"},"369693328544bf240642e143aab97590f07449c4":{"kind":"REWORK","_number":2,"created":"2021-07-28 16:44:06.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/54/802754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/54/802754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/802754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/802754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/54/802754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/54/802754/2"}}},"commit":{"parents":[{"commit":"dfa9eddb3f4b2cbed0e4fe4ec1604aac70d83bec","subject":"Show debug messages during upgrade ci job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/dfa9eddb3f4b2cbed0e4fe4ec1604aac70d83bec"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-07-28 16:36:21.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-07-28 16:44:00.000000000","tz":120},"subject":"Keep sushy-emulator state directory in /var/lib","message":"Keep sushy-emulator state directory in /var/lib\n\nBy default it\u0027s in /tmp, which means that it can be removed on reboot\nor even during normal operation (after some time).\n\nChange-Id: Ib8e55dc9909381740d79fc660e46ecd329b89283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/369693328544bf240642e143aab97590f07449c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/369693328544bf240642e143aab97590f07449c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
