)]}'
{"id":"openstack%2Fbifrost~814858","triplet_id":"openstack%2Fbifrost~master~I06cf6723361010e0a7c6eb07f2d9e570192cd9a4","project":"openstack/bifrost","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-10-25 09:57:53.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2021-10-25 08:30:08.000000000","reason":"Riccardo Pittau replied on the change"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2021-10-22 14:02:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I06cf6723361010e0a7c6eb07f2d9e570192cd9a4","subject":"Bump up Ansible to 4.x","status":"MERGED","created":"2021-10-21 06:09:22.000000000","updated":"2021-10-25 09:59:20.000000000","submitted":"2021-10-25 09:57:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"814858","meta_rev_id":"c5bac508b60ced1c27fa1e83bd76a1e44d3cc7d3","_number":814858,"virtual_id_number":814858,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-25 09:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-25 08:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-25 08:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-22 14:02:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-10-25 08:28:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-10-25 08:30:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 08:13:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-21 10:26:20.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2021-10-21 11:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 12:58:26.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2021-10-22 14:02:13.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-10-25 08:30:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"84fb9fcd9e15fd490d92bf152d936e7e445a3c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 06:09:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"957733634df1154cf661f4b32e1c1d531f07fc43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-21 08:13:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"887d4780e4498261b12810f854bf200fb0452c58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 08:52:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"67f0c7695d37b79942d8ced84d36c4710236fb2c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-21 10:26:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb5f60af76c872c8b49a568e97ce45b49bbf6d2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 11:33:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f1611c7f95546e2a59ca67e80034545 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a23e54c416c41a68e2e9646d2654ac3 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c82e1cd5edff4b4499957a8b807d69c5 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef4511e71d07455c8a0181271a12e466 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc3b62b59085492ab9ba39cf4116008e : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcda4eca2832478d8d808f84af323adc : SUCCESS in 4m 33s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/f670b7cc6e35486289b05bf53249f7dd : SUCCESS in 6m 48s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/454a5e7b2a5e4a3ebfd131425a6ff806 : SUCCESS in 45m 36s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/29e484dfa5e54dada44bd930fbe837d8 : SUCCESS in 51m 29s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d75735763594f6dab0f715a42eea4c5 : SUCCESS in 47m 11s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/195ceafd646e4d5baa0c3ba9a0f8c5aa : SUCCESS in 50m 50s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/d1301c642b1c4f778e0cec982d4b48ea : SUCCESS in 44m 33s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1fed6ee4d5643368adb7ed9d60346b5 : SUCCESS in 44m 21s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/1b68ec98a5b943c5be38fb7fa36eef4b : SUCCESS in 47m 28s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ca0655dbe3a498daf0d2307f7d90018 : SUCCESS in 1h 07m 42s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/e93703d52c15461aab2947d57ff2508f : SUCCESS in 1h 22m 27s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0cbd3c9ef0040cd8d72e63b8bf4f962 : SUCCESS in 42m 26s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/a9f436e08f6342ca9d2793f76588b923 : SUCCESS in 48m 46s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/6065f9e11e7b4c7696b47a901e8b1aa4 : SUCCESS in 45m 54s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/35afc03fc4954e7e800471f997473232 : SUCCESS in 44m 57s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/9b2f00d8171148c48a9594afbd9a7ff6 : SUCCESS in 44m 51s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65dc4f4302624c84bd771a6741c6c275 : SUCCESS in 29m 28s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/f02a0df887cd47af891a5f444cf4d614 : SUCCESS in 29m 38s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6400974102f1494990500d9a015db1c8 : SUCCESS in 15m 46s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/f0cc1f864c1a462ab0410e7fbdfff270 : SUCCESS in 22m 01s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cba4127d03a4092952592cdb6eee983 : SUCCESS in 46m 43s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/4f34e5fff0474212a890f449edcb82e1 : SUCCESS in 43m 27s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/e114acb255bd40f6b1d33b3f03f5308e : SUCCESS in 46m 34s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/9396f30896f34158a93b8d4c0412aae9 : SUCCESS in 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"947308bad8aff41caed6ffeb2e99a17e184f69d2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 11:42:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f38f7cff4179d4963176e8758f9fb2f40c29eb0","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-21 12:54:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8683602ca4ae109854b34a1b6eb125e0e7a6e58","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-10-21 12:58:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9915d08babbc952a38640e803347fca0c3eafc73","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 07:56:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"11a55777416c4a86a27ee7015efaca6c8ee95cf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 08:04:07.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f84ccc5c311bb499d40d93a30e72011716715033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 08:05:07.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8331c416bc10c9adbabd12fccfd1f328adab9d10","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-22 09:35:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b38db19c87648bb2fb90552256b2be21dcea200b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 11:20:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5fe3d54cd074753be598d07fd9e96a3 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/991334cfd43a45efa065bfe5dc31413a : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9686ebb4ff444b3a951d401437a2f210 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/928e965623e44f72bf13bea03f3935d8 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a52668434b42f09daf5e54573901c6 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af69d08d390d407085d5b2887b585155 : SUCCESS in 4m 47s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/d69f07be9ba64e4aa29163eca838c174 : SUCCESS in 6m 47s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/935c87cbc1e248d08dff1bfbbdaccb46 : SUCCESS in 44m 32s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/01e69a36eb8d4c1f84095cfadc7d617d : SUCCESS in 48m 52s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/884a76415a3d49578221e5fa77837718 : SUCCESS in 49m 35s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/8c9bb7c3b6704cb1b6b4c48f2ff7f5d5 : SUCCESS in 54m 52s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/86a79e38163c4090bb1628a169fa1299 : SUCCESS in 53m 17s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/416802fda7774e25ac431cbfe508a188 : SUCCESS in 44m 30s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/1d00de014b504cd7ba15faeb4962454c : SUCCESS in 53m 30s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77f5aeedab7f494dab8806c662c97094 : SUCCESS in 1h 07m 41s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/9d8b867c3c704c6887e753fce62dd5ae : SUCCESS in 1h 20m 47s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afc3b43fafe04c3996e3a2fe80f36707 : SUCCESS in 43m 19s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/9188025fe26f4ce494384fcf0ccbe4f0 : SUCCESS in 50m 43s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/a0a2afb7e2d54f1e9a228ed6a6aba062 : SUCCESS in 47m 35s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/4233d2b121ba4ae988efc58cbcdb9fdc : SUCCESS in 48m 00s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/4cc020a73b5f4eb6aafa51648f687963 : SUCCESS in 45m 51s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02d00ce275284c4a8354b0128e0a11dd : SUCCESS in 29m 44s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/ecd0260490ea4306b8078392bd260023 : SUCCESS in 36m 07s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84f5004c20c4482aa333cd8a00b10fd0 : SUCCESS in 20m 36s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/f69fe4d913cf4e248150a4fe41b74157 : SUCCESS in 19m 54s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30727bca849f452e8688b9c334d5aef1 : SUCCESS in 49m 06s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/498ee64f988646879f1d5de073769986 : SUCCESS in 44m 34s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/5efda8c8f0c64be18a418980d2dc2e35 : SUCCESS in 52m 18s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/c123110f7b8e4aeaaec358c9d08c2802 : SUCCESS in 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"78827b987b2ac6dfb6330669940bb872ce15d170","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-10-22 14:02:13.000000000","message":"Patch Set 4: Code-Review+2\n\nNow we need to start using new names for the actions, I guess","accounts_in_message":[],"_revision_number":4},{"id":"488c4d07805d71a2ad7302b44430c13f02ca1904","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-25 08:28:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6a1449e1a91cd6adcdeba443e30044d0ae4c7220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 08:29:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7f1672ef58a3d75908fd7f5fa841ba3ba52de210","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-25 08:30:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b380d9ae06fc1b6257be24e161ad27383cdf30be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 09:57:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60926c5951be4c63b30cffc6c1500466 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fc223ab00c14eab84334550ab681e4a : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdf30778af9a4d78abd1c9342c0385a2 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35db8bfbaca54368b482ab4ac5a78c67 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54fdda70cde145e39ecf9beb4283c1ad : SUCCESS in 4m 13s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/38924fb122754eae99518c6049307f89 : SUCCESS in 5m 33s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cce56a63eca4c5da25f92b2a8b12056 : SUCCESS in 42m 53s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/910be64653a04425b9ee27659a073cd9 : SUCCESS in 44m 09s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba5406fdc618492b8c67b2523c9ccb17 : SUCCESS in 43m 20s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/e061f53b8cb64554b404b287ecac5774 : SUCCESS in 46m 22s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/8e8fe0d8019a4e3f9499cc5a1b637269 : SUCCESS in 47m 09s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d7ef44412d648928e0ab8416b1c7575 : SUCCESS in 44m 19s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/0c9171d0e34e4ab88badf8d3f87c714b : SUCCESS in 49m 04s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c37b2ced5742409ad6ec2b20415bfb : SUCCESS in 1h 07m 02s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/53369a2519714077b69847f59fc62c48 : SUCCESS in 1h 24m 24s","accounts_in_message":[],"_revision_number":4},{"id":"d26c1a29caa2df95043ba472f8756353bd5789b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 09:57:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c5bac508b60ced1c27fa1e83bd76a1e44d3cc7d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 09:59:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42d27a398d949ff923fada0553e7d69 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11cc7e4c9929418b9891cfd66d3a45ff : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"72ad77f91470f84dcd35fdb7360a5c4e83334878","revisions":{"086fa4a7aa4f9b4e90dd93def3aea5f06fb62634":{"kind":"REWORK","_number":1,"created":"2021-10-21 06:09:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/814858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/58/814858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/58/814858/1"}}},"commit":{"parents":[{"commit":"c6d16b72ecaffcdbb9c965484de5e684f6b43cec","subject":"Merge \"Enable authentication in sushy-tools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c6d16b72ecaffcdbb9c965484de5e684f6b43cec"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 06:07:22.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 06:09:15.000000000","tz":120},"subject":"WIP: Bump up Ansible to \u003e\u003d2.9,\u003c5","message":"WIP: Bump up Ansible to \u003e\u003d2.9,\u003c5\n\nChange-Id: I06cf6723361010e0a7c6eb07f2d9e570192cd9a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/086fa4a7aa4f9b4e90dd93def3aea5f06fb62634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/086fa4a7aa4f9b4e90dd93def3aea5f06fb62634"}]},"branch":"refs/heads/master"},"805be22643b7b781fb73e3390d482bbb84b92851":{"kind":"REWORK","_number":2,"created":"2021-10-21 08:52:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/814858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/58/814858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/58/814858/2"}}},"commit":{"parents":[{"commit":"c6d16b72ecaffcdbb9c965484de5e684f6b43cec","subject":"Merge \"Enable authentication in sushy-tools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c6d16b72ecaffcdbb9c965484de5e684f6b43cec"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 06:07:22.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 08:49:44.000000000","tz":120},"subject":"Bump up Ansible to \u003e\u003d2.10,\u003c5","message":"Bump up Ansible to \u003e\u003d2.10,\u003c5\n\nAs per [1] and [2] - Ansible 2.9 is not really maintained anymore\n(only security bug fixes will be applied).\n\n[1]: https://docs.ansible.com/ansible/devel/reference_appendices/release_and_maintenance.html#ansible-community-package-release-cycle\n[2]: https://docs.ansible.com/ansible/devel/reference_appendices/release_and_maintenance.html#ansible-core-release-cycle\n\nChange-Id: I06cf6723361010e0a7c6eb07f2d9e570192cd9a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/805be22643b7b781fb73e3390d482bbb84b92851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/805be22643b7b781fb73e3390d482bbb84b92851"}]},"branch":"refs/heads/master"},"6775ed71f1faa096fd77ebc0aa2637761a724754":{"kind":"REWORK","_number":3,"created":"2021-10-22 08:04:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/814858/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/58/814858/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/58/814858/3"}}},"commit":{"parents":[{"commit":"c6d16b72ecaffcdbb9c965484de5e684f6b43cec","subject":"Merge \"Enable authentication in sushy-tools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c6d16b72ecaffcdbb9c965484de5e684f6b43cec"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 06:07:22.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-22 08:04:03.000000000","tz":0},"subject":"Bump up Ansible to 4.x","message":"Bump up Ansible to 4.x\n\nAs per [1] and [2] - Ansible versions prior to 4 are not maintained.\n\n[1]: https://docs.ansible.com/ansible/devel/reference_appendices/release_and_maintenance.html#ansible-community-package-release-cycle\n[2]: https://docs.ansible.com/ansible/devel/reference_appendices/release_and_maintenance.html#ansible-core-release-cycle\n\nChange-Id: I06cf6723361010e0a7c6eb07f2d9e570192cd9a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6775ed71f1faa096fd77ebc0aa2637761a724754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6775ed71f1faa096fd77ebc0aa2637761a724754"}]},"branch":"refs/heads/master"},"72ad77f91470f84dcd35fdb7360a5c4e83334878":{"kind":"REWORK","_number":4,"created":"2021-10-22 08:05:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/814858/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/58/814858/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/58/814858/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/58/814858/4"}}},"commit":{"parents":[{"commit":"c6d16b72ecaffcdbb9c965484de5e684f6b43cec","subject":"Merge \"Enable authentication in sushy-tools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c6d16b72ecaffcdbb9c965484de5e684f6b43cec"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-21 06:07:22.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-22 08:05:07.000000000","tz":0},"subject":"Bump up Ansible to 4.x","message":"Bump up Ansible to 4.x\n\nAs per [1] and [2] - Ansible versions prior to 4 are not maintained.\n\n[1]: https://docs.ansible.com/ansible/devel/reference_appendices/release_and_maintenance.html#ansible-community-package-release-cycle\n[2]: https://docs.ansible.com/ansible/devel/reference_appendices/release_and_maintenance.html#ansible-core-release-cycle\n\nChange-Id: I06cf6723361010e0a7c6eb07f2d9e570192cd9a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/72ad77f91470f84dcd35fdb7360a5c4e83334878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/72ad77f91470f84dcd35fdb7360a5c4e83334878"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
