)]}'
{"id":"openstack%2Fbifrost~819057","triplet_id":"openstack%2Fbifrost~master~Icf40c0ded39b2afa9a667425971faa61f260c9c8","project":"openstack/bifrost","branch":"master","topic":"remove-job","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2021-12-06 11:37:26.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Icf40c0ded39b2afa9a667425971faa61f260c9c8","subject":"Remove unused integration job","status":"MERGED","created":"2021-11-24 06:52:47.000000000","updated":"2021-12-06 11:41:58.000000000","submitted":"2021-12-06 11:37:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819057-remove-job","meta_rev_id":"c139f7a55eed9aca8cb74be3f60ac8d2dc492b76","_number":819057,"virtual_id_number":819057,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-06 11:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":2,"date":"2021-12-06 10:45:14.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-24 11:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":1,"date":"2021-12-06 10:45:14.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-24 08:46:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-24 11:32:05.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-12-06 10:45:14.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"}],"messages":[{"id":"93fbaf231a024437b7e86ebe8306f3a146f21fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-11-24 06:52:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6925803f5e836e5cee9df7dc5da79416fba6717","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 08:46:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a77d789be84449a6bd91367c8ed62891 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7c8f07da41547c9abde68e4f0156cd0 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f2a0cced0e641db9eb194067dc605b4 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aca414d68434228b5ddb228ed88b38f : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b192acba7799425b937adebc50029c5f : SUCCESS in 5m 09s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/6384f124119443e2a1fc9ce9a5f78241 : SUCCESS in 5m 49s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8114dcd1e7a34c8f8fa47b74ce49a1af : SUCCESS in 24m 17s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/27caa56639264e2984214367b8dd282f : SUCCESS in 34m 11s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/045847bdde494fc1a9b66bfdc0e65745 : SUCCESS in 36m 12s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/a194834b528448469706dd01360b78c7 : SUCCESS in 43m 07s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/aea0f8ee75f04f089b40f3f9f41d57f5 : SUCCESS in 23m 11s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99e6aa0edcce44bd8016684b618eb919 : SUCCESS in 33m 45s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/5599fb60105a40719ee70650ab932d31 : SUCCESS in 34m 02s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e081793b88d435b9449df8dd918cd53 : SUCCESS in 1h 02m 46s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/d58d4398e9b64314b1c5d0d3f4dcbd37 : SUCCESS in 1h 15m 16s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4d8a9ee0b1f4daabd64d8f62f3028b2 : SUCCESS in 30m 38s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/0615c10c2bb9422aada3dff6402b1b63 : SUCCESS in 32m 56s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/ae7ad0f672b4414db2a8faff6f0d904c : SUCCESS in 26m 04s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/63f6f6508d3443fa8fe179a8aa9d651c : SUCCESS in 28m 38s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/46cd35a5f0494c2db2066006eb9a802f : SUCCESS in 35m 32s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd0705660f08411d86b9d88082e0079f : SUCCESS in 38m 34s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/8e771f7027c1477093526841fdaf548d : SUCCESS in 38m 33s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5380ae58b6e14dee9fdb120b8f10f540 : FAILURE in 14m 49s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/255c5fc4476a4e71b92d04beca3f8334 : FAILURE in 18m 17s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e694ed54a634dfea9be545063241892 : SUCCESS in 31m 33s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/6252544586ec44a29212993a711fb3d2 : SUCCESS in 28m 53s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/656e03b0084e410584ffc75b1c75b60a : SUCCESS in 31m 41s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/1292382a46f54c55a1b043775a02c090 : SUCCESS in 40m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f2cba23c64bffe2b68dc137392c61fbe333fce7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-24 11:32:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"005f403e93a3388696689271cc2fcf39a31347d6","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-11-29 15:17:03.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"86acbef69d8e4d14d92f68b6fcd035741025a7a4","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-12-06 10:45:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3b9b80c723cc3731595b6cee259f895a4e8fb2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 10:46:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"787ebffa83af2c20f9a30c1b79a879f8df9478e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:37:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52f5dad050304288b3dbc195db908e8a : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d726eeedd64f4eef86f44cc7dca5044d : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c76e7777db13454aa33dd89dd23f3ab1 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803816f4b1de47bbae40fa756d0cdc73 : SUCCESS in 7m 36s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/3a814b71fcc8410798a02889d5e59958 : SUCCESS in 7m 15s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2dc082d7eee408590eea953677e84cd : SUCCESS in 27m 36s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/d383b0cbe4074fe888aed2a810d21508 : SUCCESS in 36m 23s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d097c44f68d4bbdab178fdce3495293 : SUCCESS in 28m 38s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/76d90cd812934b17a456c9aabe84a693 : SUCCESS in 32m 22s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/cf08312ec7a942538f1eef1326abd42b : SUCCESS in 28m 00s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec68275dc2f1409e8676729867ff6b2f : SUCCESS in 30m 54s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/a3d3a5e8609e461e94cc0d367e78e23b : SUCCESS in 34m 53s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84ad52d2f6364341aba6bf2290f85e57 : SUCCESS in 49m 51s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/4fa14962efc04593a9cd6dde7b09a673 : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":1},{"id":"40992d1d0afdb6b8117771d2c91a49110bb1e88f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:37:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c139f7a55eed9aca8cb74be3f60ac8d2dc492b76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:41:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f202623791fc42de857d166027e29d68 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec042fa3b864f51661864675d632e66a6c8198c3","revisions":{"ec042fa3b864f51661864675d632e66a6c8198c3":{"kind":"REWORK","_number":1,"created":"2021-11-24 06:52:47.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/57/819057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/57/819057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/57/819057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/57/819057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/57/819057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/57/819057/1"}}},"commit":{"parents":[{"commit":"6d9cc8935911446a5c77ce71791ad98a7984169d","subject":"Merge \"Refresh supported operating systems list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6d9cc8935911446a5c77ce71791ad98a7984169d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-11-24 06:51:10.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-11-24 06:51:54.000000000","tz":60},"subject":"Remove unused integration job","message":"Remove unused integration job\n\nThe ansible-collections-openstack project has moved to\nbifrost-integration-tinyipa-keystone-ubuntu-focal since a while.\nNo other projects are using bifrost-integration-tinyipa-keystone-ubuntu-bionic\n\nChange-Id: Icf40c0ded39b2afa9a667425971faa61f260c9c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/ec042fa3b864f51661864675d632e66a6c8198c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/ec042fa3b864f51661864675d632e66a6c8198c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
