)]}'
{"id":"openstack%2Fbifrost~827128","triplet_id":"openstack%2Fbifrost~master~I40b3e863f78e185c39a20f4254b68b8140eea9e0","project":"openstack/bifrost","branch":"master","topic":"unix-socket","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-03-11 15:18:19.000000000","reason":"Removed by Dmitry Tantsur by clicking the attention icon"}},"hashtags":[],"change_id":"I40b3e863f78e185c39a20f4254b68b8140eea9e0","subject":"Make ironic and inspector listen on Unix sockets","status":"MERGED","created":"2022-01-31 15:20:45.000000000","updated":"2022-03-11 15:54:24.000000000","submitted":"2022-03-11 15:46:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827128-unix-socket","meta_rev_id":"06d4481c611f7344259e414b4c02868dc9a77270","_number":827128,"virtual_id_number":827128,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 15:46:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-02-24 10:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-11 14:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-11 14:13:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-31 16:15:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 10:42:00.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-03-11 14:13:20.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"84c004b57fb5633842edde99ba4eac4a576c17aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-01-31 15:20:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af79c493811361718e9bfbd515678e86af39bfdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-01-31 15:23:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cf092ee537e1ef4e9fe9ecca124a8ba81e97a069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 16:15:33.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"944102c23e4b84603d77aa093c27b4d9e7828aae","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-01-31 16:15:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14ee6bf041b97c6d8709134b0144c5bcb48b192e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 17:58:38.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"9d518ab2d9936a03bc19b9284f30595ea8122124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-01-31 17:59:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ceb200214e4b6fbb473bac19dbc7731665c9dc7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 19:39:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adae337895054aea90104c549e8de78f : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d27e57e5ecc4094b07fefc728e8cf87 : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d52c0068c41744fc9de0c10b1070f144 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede8f1898ac041ddbb0bf9b9c91ddc0d : SUCCESS in 6m 28s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/3ea2450ecc0e458d85f8f98439830c29 : SUCCESS in 7m 41s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/7ac3737994f647aaa7e69d3e707d643b : SUCCESS in 26m 26s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bebd0722b29e496fa2a745b45bc3ecb2 : SUCCESS in 29m 24s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/460c41cbcd5a4defa6bc252d315427f5 : SUCCESS in 34m 34s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/222562ff988548fc9ad46624da34d210 : FAILURE in 17m 46s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/eeafa4f3159c4e46851fb19e232cc8a3 : FAILURE in 23m 11s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/5a5e78679c8d48618fed901950c9abd7 : FAILURE in 22m 26s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/1ae9fa7db0114d8a83e423971b8b9584 : SUCCESS in 26m 26s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf0d2b34be6b4bb09ca65f76a72feeae : FAILURE in 15m 18s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/334e8416bf8a4830a7d44d47cd273221 : SUCCESS in 33m 35s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd841c407c854895a7f79ac7f44dfa7e : SUCCESS in 46m 57s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/a35b84b4111942edb32c1ef5afedf965 : SUCCESS in 44m 22s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64652f3c513140ee964af69f2dfcc64b : SUCCESS in 27m 39s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/02e76278841a4bbdb1f109b41ba5ce7a : SUCCESS in 30m 26s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/5a2709de587c407eab748c153e44e08f : SUCCESS in 36m 20s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/432d2c997b694ffd8aa32ba5280b794c : SUCCESS in 25m 49s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/456797d297004cc6a1e39fccdc77c6ad : SUCCESS in 35m 27s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a340eea860fc465a9cc9e47bd7723d1e : SUCCESS in 35m 33s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/a6f87b789af04bf29c7d7d52296de734 : SUCCESS in 41m 34s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d40a549460f64085aa704cf9a7322b19 : FAILURE in 37m 06s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/a6f61bf4195e4148b652c60d76f71404 : FAILURE in 39m 18s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b03137541a5b47aa8428f4e156aca291 : SUCCESS in 31m 08s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/bf0a1ee1caf3413d947cbd361c46b602 : SUCCESS in 28m 25s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/1407c16b5345475698485239e27d3335 : FAILURE in 20m 01s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/f1513016a34b480fb7a13bcf35e8d1a1 : FAILURE in 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4cfae12812cfd406fdf7ac22b4de0bdb141b0d22","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-01 09:24:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7be60e7c89f39532ab32857d53e70de927a94938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 09:30:48.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"aab346742d20c9299d6fb990af154e870c71cf09","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-01 09:39:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7391a5be35e18ddf527a111324e608abed8b1e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 10:29:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33702d779cba4902a19cd4d959414b49 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f064f57b6ee47f8b7b1e6e478c5f5ce : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46d83a95789742d39765ce03b5620242 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c46abe436714d2d92ca43ec6b887cce : SUCCESS in 5m 22s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/544483ea64054338b35b7004317d4656 : SUCCESS in 6m 15s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/fe195f14a85b4b43bb4b7a2040cbf51c : SUCCESS in 26m 30s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e48c233f229c4aee8213d3b93b5012d9 : SUCCESS in 20m 49s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/328ee4cf479448dd924062dc0d7ce281 : SUCCESS in 25m 08s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7863da4004be456685d549f05e2a6a23 : SUCCESS in 26m 17s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/f2cfa69e7b87410ebec8e34cf3ab5efb : SUCCESS in 28m 32s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/7074d65c316347bcad57ab0b5f82abbd : SUCCESS in 27m 30s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/f8c2427683ec41bf803708a8e717ec06 : SUCCESS in 23m 08s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d35aec0148284c84976e1255b908cdae : SUCCESS in 23m 37s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/0268b74d9fc94ba4b2ec50fa820b47be : SUCCESS in 28m 11s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df07827974c040d98fca3f6389778fe9 : SUCCESS in 31m 30s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/f1c5669b386849aca4fc018fb369aff4 : SUCCESS in 43m 35s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44f79d761c7e46fbab0d14371ea96613 : SUCCESS in 28m 21s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/091e7cdad08149ce9c454ad7572bcfd4 : SUCCESS in 32m 53s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/854a69b47fe4461c804830239ae3d8b6 : SUCCESS in 32m 56s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/7d4af0e73053428fb9528f0ab9927f17 : SUCCESS in 28m 47s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/589908f973234a33b9f324a29cc36484 : SUCCESS in 34m 22s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6755c8fd00a84c5980806b1421d6d80f : SUCCESS in 33m 22s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/2f2cc4674df746a9b06604d7ffb9101a : SUCCESS in 42m 46s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/995747a5f9fb4d2d8806056fa57f0a41 : FAILURE in 38m 04s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/6d4ce060a1cf4b6f82f854f116854dd0 : FAILURE in 43m 49s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5afeb9cc59c74ad397af6a124562b41b : SUCCESS in 41m 45s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/c786664e0b5e4bffba878dd27035fa28 : SUCCESS in 28m 48s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/14ac95fdc1f54f8a8c83fde24bbb4866 : SUCCESS in 33m 21s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/1f9b7b42a47d4dd6a56834a72cea173d : SUCCESS in 36m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4c30c5b06ce656cece04b8d925018e104c398d28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-09 10:01:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6b2420d0b3836351862ed66934843bfabae23d67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-09 10:07:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a73a713de573cdfbcaa3a890569fe09a5098dea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 10:33:33.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"5be42c14e978a17c636ec646d1ba0c284e82aa72","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-09 10:34:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5b1eb71d03d18bddcf240b755b38660b0acc5e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 11:41:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce025df31b7241c3ad03c908cf7e667f : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7b3fdee6ff84b1f95bbb4e80e732421 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/193103814ac14322bb0b5a73ad06c03b : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d2ac479b3b452199eaf1bc55dc59b4 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a032a1b2edd4d7e9a88e55d68dad888 : SUCCESS in 5m 55s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/15c167b76c6c4843b5161e005f323fec : SUCCESS in 8m 43s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/a20ed0771add440692244e3458e46903 : SUCCESS in 37m 05s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4da1c0d538c4c128c6c7e40b0696544 : SUCCESS in 27m 49s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/40d3ec30888f4ba4885316e16f1244ad : SUCCESS in 40m 40s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ca161467a2a4d0ebbd89e522d752de0 : SUCCESS in 28m 55s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/c7b072288f9c46e28893c78e71de9da7 : SUCCESS in 34m 21s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/6b3dfcbd9a5d4c24ab8f872052b50e66 : SUCCESS in 35m 49s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/d22cb0f7a5c640069597782f278e30e8 : SUCCESS in 25m 11s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d5e5922e428409899016aac0d9696f9 : SUCCESS in 27m 43s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/cd78ee6b8b9b42b199fd43192fef90e1 : SUCCESS in 28m 42s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa7d1afb0d4c4e01a57db73402c91fc7 : SUCCESS in 48m 08s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/651c068cb8b943ae92efc00d3505a368 : SUCCESS in 58m 39s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34bc6f00fcc2463bb3b01d86a5711028 : SUCCESS in 32m 18s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/d0d70f39a23d452f9bc05676f36e3ffc : SUCCESS in 29m 41s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/a3e453cf98c24d2e8d41556978b339c8 : SUCCESS in 32m 17s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/db62298d08f34ff6b1a7245a1c05f016 : SUCCESS in 25m 27s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/b5234fc4816b49128c2d8b9b3a5926a1 : SUCCESS in 34m 07s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04cf089fbafc4fb89d9c9b90a5ab1640 : SUCCESS in 32m 21s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/f29f66d2b36e453cbfd987c36485896e : SUCCESS in 36m 36s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/153387fc6f9c4f4eb61399ac93b2a34c : FAILURE in 38m 30s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/7774afa119604c82b35dbe39a6f4346b : FAILURE in 39m 59s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fbcae0bf50746799d34e2fbdf605854 : SUCCESS in 28m 59s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/e6674e3123ed45d7b92fd81af523afcb : SUCCESS in 28m 18s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/6ce3e7a5bb0f4da794f46ec99504fd02 : SUCCESS in 28m 50s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/131bd5db153e4dd5a95062f548309ef4 : SUCCESS in 38m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bd32111733a60c454f0b0e5c9d79a111444eb007","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-02-24 10:42:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2cf125b2eacd75487a195b8302f626d0f4b026b0","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-03-11 14:13:20.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e90c90ae51b7497e2e267b4abeed4aadce38db82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 14:14:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7dfbabb18dd9ed1f4472dd3d440f857fcaaeb3fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 15:46:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036f947a427742a691c93267751c89cc : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a376e057e56c40b7bef6735b93d5ddd9 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5858ea10ef9e4599bf1d3c455d5688ee : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2c588a62f440a48e900e8727a92870 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04de1d3a58a0435a8688ad314bc25913 : SUCCESS in 5m 21s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/818a7fb78f0d4a5d9a7a4ecd18534163 : SUCCESS in 8m 54s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/409dfdfef66b48cfaca2e96602e1d55d : SUCCESS in 27m 38s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/556f1c0341454c53b66c5162c99c249e : SUCCESS in 28m 26s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/4ef9c85decb24364aeeff442c96d9db1 : SUCCESS in 31m 44s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14ddb5880c764ac28ba304176a4e2ff2 : SUCCESS in 29m 01s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/1a566d7d262c40d0a4826eaed0d3367d : SUCCESS in 37m 13s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/487a67e5b32a4e6dac42e9a0c4d46cac : SUCCESS in 31m 11s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/675296eb128a44e19cb05b2d6c884805 : SUCCESS in 27m 04s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c875d21a9806467a88f216b9a69634de : SUCCESS in 27m 00s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/edbcf0fce0b446e7a707a40a4e422b92 : SUCCESS in 32m 58s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/25a6e20891d24281b59c10bf1570635a : SUCCESS in 29m 41s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15ece17df9424af99c45683d61d22d04 : SUCCESS in 46m 58s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/581e7f03a71143498d553fe3281b39ee : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":5},{"id":"f6bef2da1744da0297b5fdc455e1fbcbdc349f70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 15:46:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"06d4481c611f7344259e414b4c02868dc9a77270","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 15:54:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/127a889f2a0d4d96838f48d58fba49dc : SUCCESS in 2m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/888fce2851af496b860e3812b447d337 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c0f901c1b4aa251cb7e0967be836fb369855629a","revisions":{"78e186903808567d31b476515ebaf1351238a0cc":{"kind":"REWORK","_number":1,"created":"2022-01-31 15:20:45.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/28/827128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/28/827128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/28/827128/1"}}},"commit":{"parents":[{"commit":"9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd","subject":"Merge \"Make virtual environment owned by root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-01-31 15:19:55.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-01-31 15:19:55.000000000","tz":60},"subject":"Make ironic-inspector listen on a Unix socket","message":"Make ironic-inspector listen on a Unix socket\n\nThis way we don\u0027t expose the plain text endpoint to anyone even locally.\n\nChange-Id: I40b3e863f78e185c39a20f4254b68b8140eea9e0\nDepends-On: https://review.opendev.org/c/openstack/ironic-inspector/+/827124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/78e186903808567d31b476515ebaf1351238a0cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/78e186903808567d31b476515ebaf1351238a0cc"}]},"branch":"refs/heads/master"},"9f527c5f96f3178fd0db413dd404dbea5cb5e3b2":{"kind":"REWORK","_number":2,"created":"2022-01-31 15:23:35.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/28/827128/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/28/827128/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/28/827128/2"}}},"commit":{"parents":[{"commit":"9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd","subject":"Merge \"Make virtual environment owned by root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-01-31 15:19:55.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-01-31 15:23:29.000000000","tz":60},"subject":"Make ironic-inspector listen on a Unix socket","message":"Make ironic-inspector listen on a Unix socket\n\nThis way we don\u0027t expose the plain text endpoint to anyone even locally.\n\nChange-Id: I40b3e863f78e185c39a20f4254b68b8140eea9e0\nDepends-On: https://review.opendev.org/c/openstack/ironic-inspector/+/827124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9f527c5f96f3178fd0db413dd404dbea5cb5e3b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9f527c5f96f3178fd0db413dd404dbea5cb5e3b2"}]},"branch":"refs/heads/master"},"bb6d444b0f0215adef081ca133f92e84c0931a8b":{"kind":"REWORK","_number":3,"created":"2022-01-31 17:59:38.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/28/827128/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/28/827128/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/28/827128/3"}}},"commit":{"parents":[{"commit":"9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd","subject":"Merge \"Make virtual environment owned by root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-01-31 15:19:55.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-01-31 17:59:12.000000000","tz":60},"subject":"Make ironic-inspector listen on a Unix socket","message":"Make ironic-inspector listen on a Unix socket\n\nThis way we don\u0027t expose the plain text endpoint to anyone even locally.\n\nChange-Id: I40b3e863f78e185c39a20f4254b68b8140eea9e0\nDepends-On: https://review.opendev.org/c/openstack/ironic-inspector/+/827124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/bb6d444b0f0215adef081ca133f92e84c0931a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/bb6d444b0f0215adef081ca133f92e84c0931a8b"}]},"branch":"refs/heads/master"},"8d5174e0a05aed4ef49597f61db3307fb6c9f344":{"kind":"REWORK","_number":4,"created":"2022-02-09 10:01:46.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/28/827128/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/28/827128/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/28/827128/4"}}},"commit":{"parents":[{"commit":"9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd","subject":"Merge \"Make virtual environment owned by root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-01-31 15:19:55.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-09 10:01:05.000000000","tz":60},"subject":"Make ironic-inspector listen on a Unix socket","message":"Make ironic-inspector listen on a Unix socket\n\nThis way we don\u0027t expose the plain text endpoint to anyone even locally.\n\nChange-Id: I40b3e863f78e185c39a20f4254b68b8140eea9e0\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/828508\nDepends-On: https://review.opendev.org/c/openstack/ironic-inspector/+/827124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/8d5174e0a05aed4ef49597f61db3307fb6c9f344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/8d5174e0a05aed4ef49597f61db3307fb6c9f344"}]},"branch":"refs/heads/master"},"c0f901c1b4aa251cb7e0967be836fb369855629a":{"kind":"REWORK","_number":5,"created":"2022-02-09 10:07:11.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/28/827128/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/28/827128/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/28/827128/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/28/827128/5"}}},"commit":{"parents":[{"commit":"9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd","subject":"Merge \"Make virtual environment owned by root\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9c2a1790fc3ff8a5114c2423f59e3ef22763d2dd"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-01-31 15:19:55.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-09 10:06:41.000000000","tz":60},"subject":"Make ironic and inspector listen on Unix sockets","message":"Make ironic and inspector listen on Unix sockets\n\nThis way we don\u0027t expose the plain text endpoints to anyone even locally.\n\nChange-Id: I40b3e863f78e185c39a20f4254b68b8140eea9e0\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/828508\nDepends-On: https://review.opendev.org/c/openstack/ironic-inspector/+/827124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c0f901c1b4aa251cb7e0967be836fb369855629a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c0f901c1b4aa251cb7e0967be836fb369855629a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
