)]}'
{"id":"openstack%2Fbifrost~827687","triplet_id":"openstack%2Fbifrost~master~I282f690214917eb4c17adfefc70a2ecddb7144fe","project":"openstack/bifrost","branch":"master","topic":"upgrade","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-02-08 21:29:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I282f690214917eb4c17adfefc70a2ecddb7144fe","subject":"CI: properly report failures in the upgrade job","status":"MERGED","created":"2022-02-03 15:11:02.000000000","updated":"2022-02-09 08:14:06.000000000","submitted":"2022-02-08 21:29:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"827687-upgrade","meta_rev_id":"017b3bc3fdc2452ad08cea6446de4d095be0cf1b","_number":827687,"virtual_id_number":827687,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 21:29:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-02-08 14:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2022-02-08 19:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2022-02-08 19:37:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 16:09:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 08:47:02.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-02-08 19:37:05.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"ba23fc6633dbbe18525123460ef25653224b26d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-03 15:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"827a6f69770303ed29d3cb04adfe736addcabf57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:09:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f63ab58cdfa24e9ca2e5e8771304c67c : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcdfa1ebf73e43bfa45a8dc276505818 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ff6c2db62d141f5bfcb294d6fca9e4c : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151ab183410a49fbb1aa92a3ae9808a6 : SUCCESS in 7m 11s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/af41d388d48f486cac77a0b28cbb7341 : SUCCESS in 7m 39s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/be348c02953e4adda1c9dd563ea30049 : SUCCESS in 31m 27s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f726d8d576764bff85917e3cd4658a6c : SUCCESS in 30m 10s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/6387034833fa4158b744c19251612059 : SUCCESS in 36m 05s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98dd61fdadaf44e7b19800f74e772961 : SUCCESS in 33m 00s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/9817e5b248234b6ea59e6149b1de4b4b : SUCCESS in 38m 57s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/cfc059d994d641eca02a0e35c53003a6 : SUCCESS in 35m 42s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/aefb77deb19c44bb94255af4a759aed0 : SUCCESS in 28m 24s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43405f4d16f6400a8a4d0e6d0bc788b4 : SUCCESS in 31m 27s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/dbaf349947984a12953ef349bda666e4 : SUCCESS in 33m 08s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84bafbf5e0a04459b403abba19935521 : SUCCESS in 52m 38s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/6df9b1afee3a4277b0b57630ec552016 : FAILURE in 14m 42s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3b65163d7e840ff9264bac3889dc12b : SUCCESS in 28m 07s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/aed9964e9f2b49968992770258048b67 : SUCCESS in 33m 28s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/4974b17f3c4b48688cb9672f3675d2d6 : SUCCESS in 37m 11s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/ca144cd803cf480a8a369095db693052 : SUCCESS in 29m 32s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/72c6373918324b959192aba06d14d047 : SUCCESS in 34m 55s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c70076c68d7b48adb6ecf07ddd8307ce : SUCCESS in 36m 11s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/e40ff03a61bf43e697ddb73b1776f727 : SUCCESS in 46m 42s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97fd3aafc6544ded84527db385fcb5ee : FAILURE in 37m 29s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/d5caec063d72472d8978e2599824923e : FAILURE in 40m 52s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f9f135846fa42b8b4bdca50c82fdc6d : SUCCESS in 33m 50s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/5ec7bbfbe0ea44ffa559e4e7bc1bebca : SUCCESS in 36m 53s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/102e55ac4fb2491e8f760ed52a113dff : SUCCESS in 32m 03s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/c97d74901c6e491fbbd4b1fdbad65f4e : SUCCESS in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e508280a947a9f5898c5267ab80ef908e0d24df6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-02-04 08:47:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a3bbc658947cab780e99c4ad86c087b6373e6b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:33:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3edf2b6432e3470ebbf00a15d6b60432 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6c18ce752024619a5173c9d1816658f : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1044f5ab5c942949986477426068670 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9a3eae2cbc41fca7969d06cb9e49af : SUCCESS in 6m 05s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/694ad011fdda4fc8832fc9cffec3dad9 : SUCCESS in 7m 27s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/cd9b94dbdb114445b28713921668e50b : SUCCESS in 32m 22s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3e8dff578b9411db7e61bb91f899128 : SUCCESS in 28m 59s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/38ee0c13875448868f737fbed5e0801a : SUCCESS in 32m 11s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a7f8fe7a6874047b8c99aae31b7987a : SUCCESS in 32m 09s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/719301c27306460a8d74ae78c1bec905 : SUCCESS in 39m 43s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/89dcf4c9e81e43b384f99146ecec0d99 : SUCCESS in 36m 00s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/635e131a03f94e5e81e404d705fb273a : SUCCESS in 22m 56s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73bee15751b94adb948abceee925d48a : SUCCESS in 28m 47s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/4ef32ca71b2440cdae18e91a13c35a07 : SUCCESS in 38m 08s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3060bd079b2442ccab45e83407f1857a : FAILURE in 40m 08s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/71273dd4e1834906b0b714b073fec2f3 : FAILURE in 12m 42s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a287371a05043ceb974b89f9cfd3dc1 : SUCCESS in 29m 29s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/54f924fbcf0644318a9e44b8236743bb : SUCCESS in 31m 48s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/2334873ae66d4d3bbb08e654f0bed3b5 : FAILURE in 8m 02s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/8a747ee301fd48bba0cf885ae546e4fe : SUCCESS in 27m 39s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/5d8453eed96a444e8b9478141d7f9786 : SUCCESS in 36m 42s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbb74de1ddc047d4a0040c4422b52939 : SUCCESS in 34m 36s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/a203054c020a47f0b769a3821e2a7253 : SUCCESS in 38m 29s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e55bed8ba1af448a9ca212fb3238dc05 : FAILURE in 38m 06s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/9922f5aa26234bc3b1920bf44d46cc6f : FAILURE in 41m 17s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5db120dbb8f84669bab91f974d18ef3b : SUCCESS in 30m 37s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/f447113ac0dc480ebae836e0304a12e1 : SUCCESS in 29m 22s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/01ecd71701f1416eb95aa9a45899d800 : SUCCESS in 31m 13s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/c89bf5ccb90e4b6d89c1fb86471eb58e : SUCCESS in 36m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f1016e8eff5c819f186cdae7204727115e3058f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-04 09:49:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c629a567de4748ad2e1dc415c2b1a3407ef602a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-08 08:06:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"13a089d168c27f12c52d4afea6e2abf33a2a4f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 09:14:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2325ab692f24663af88ab8f327c5677 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0964c06d0d84178a0b6b750424b70ce : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef2e01703daa43f68e655f16b0eae8f8 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/384890690fe04537a05a4fe7076a0820 : SUCCESS in 5m 29s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/03cf86e492b24077a9593c877a30c122 : SUCCESS in 7m 04s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/b8c925ea54eb4f989d791a5ffba36d5e : SUCCESS in 30m 35s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75777d68a4a2496281df1558df93502e : SUCCESS in 28m 20s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/18bcf1e96a64461d95e3af398451fe5f : SUCCESS in 34m 32s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b163e9a62f4e42d1bfd98ae6aa690b3b : SUCCESS in 30m 20s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/0ffd639d2b864a149bd17a04c2562a48 : SUCCESS in 38m 19s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/ce95bfe7aba449a6af80a68d4639ec43 : SUCCESS in 33m 01s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/a4c62e3fde544ab8887bff19173a959c : SUCCESS in 27m 04s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/625aa8b3b11343cda038c290634e0e1c : SUCCESS in 27m 37s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/73865d99e3b3426587282179988fa09b : SUCCESS in 32m 31s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80b7586c2b8844f0bda876fc31c6fafe : FAILURE in 38m 37s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/02f8ad7c54f440439e3e05f26abc4f8b : SUCCESS in 1h 06m 02s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42b87c55c60d4a1ca61009062e3ac888 : SUCCESS in 28m 45s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/23f6e7963215488ca301c578b7517cfe : SUCCESS in 30m 32s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/78a1e529c9ee4c58a5a97201e7a09165 : SUCCESS in 35m 25s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/7d65d3607b054b35a2d28bb86fb70393 : SUCCESS in 25m 13s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/c0229962efc449af83a97bd41b72f782 : SUCCESS in 34m 27s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abdf850673b44d60af709cb86c326a10 : SUCCESS in 31m 12s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/9f722428f6044be79bf5725393bd8b9d : SUCCESS in 39m 21s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b85b1af9c4f431888f8c432d2bb4e04 : FAILURE in 36m 31s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/76116a87c0af45a69b3485e145cac66e : FAILURE in 40m 11s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f891cc841854446d80dd888ece00006d : SUCCESS in 33m 13s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/d0b9f99e061949d383d0e38af399eea2 : SUCCESS in 28m 36s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/cc632ea4e7084fc89e6e67d6801d7c32 : SUCCESS in 28m 23s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/2a8c09a3f1ee457d95c76ff575402c52 : SUCCESS in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"07e2bd59d711d9b465235b4b3fb0b1d4f27290bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-08 09:17:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5947ec487eabd0caad86edc9b7c18b0b974f1320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 10:34:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df19972ed80b4012bbaf56b022b82aaf : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/324f8fef49894f33b88654fbde9bcb15 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1878de6720ce4e44b2f81fe3d27f6e5d : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/320d63f1cdc44c6f946181e5f36efa22 : SUCCESS in 5m 09s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/79b770791f5544e5a4efb2f61d49e061 : SUCCESS in 6m 14s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/e111abeaefbb407e98916ef8ca8809ad : SUCCESS in 32m 41s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/095a05d63ebe4d7285ae76584bb59c15 : SUCCESS in 28m 35s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/13873aaae170490c8a60992cef30ecb8 : SUCCESS in 35m 39s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43e17cd041cb4af1aa49dd64312c821f : SUCCESS in 33m 27s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/917c677960b64f8ab2ec7b47c7efb2b0 : SUCCESS in 27m 46s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/5fa84ed8d2194ad0964209d0295b32c6 : SUCCESS in 28m 13s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/53da3efb9798495c8d3e6bff73698349 : SUCCESS in 27m 53s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0632e219cf44ef7ac979c67f6ebeddf : SUCCESS in 24m 18s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/82c19f2c81b04bd79561645ad47bcfbb : SUCCESS in 34m 43s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b74103fffae84582a66abb3857648b6f : SUCCESS in 50m 05s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/c65d093a732448818e36f658601d211d : SUCCESS in 1h 11m 15s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f13904679dd14555be58f5579bb5fbe5 : SUCCESS in 28m 51s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/2511658339eb415dab04433feddc0d6e : SUCCESS in 31m 26s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/4a43e463cec54cbc9cde1b56b730e1ee : SUCCESS in 34m 37s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/5cb1659bf8554b17b85c605bb1faaea2 : SUCCESS in 32m 33s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/9a0c77b842104968bae4020bb3805c31 : SUCCESS in 34m 58s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab063e2d55c1406b84554a85928aa21a : SUCCESS in 33m 19s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/b43e1d9ab14346a1a98d6c6905b7c0d2 : SUCCESS in 33m 57s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbf2f5be074e4739a98c0b9cf948ca30 : FAILURE in 36m 23s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/f45a0cf660c74e2991c5f1b890ee1bb0 : FAILURE in 36m 18s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf0c89026c0d47b1a92762ceae65b403 : SUCCESS in 31m 22s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/29916065ad944bd59e586d7c15e05568 : SUCCESS in 28m 47s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/d7a0d99ed8874b409e7906c4ac19b69a : SUCCESS in 30m 25s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/151bb401f9384490b2cc84b379704b0d : SUCCESS in 44m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"28794e91bc25433fe2e6f9791a02fdf926b5656a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-02-08 14:13:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a2879884e731051b91eff3a16bffb76585ced5e0","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-02-08 19:37:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0820ea6e34272a5fdc89ce56be31848e1e93dd36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 20:10:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1b0afa1df7f660ee16f3648967fb50f4b6a308da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 21:29:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2aba335be846959a0d06c4c44665ae : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee05c5378941421bbd190bdd539fd544 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8be474a37ab74349b0f2c1813e94eb82 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de94413f679470b98524ee2ceac5b32 : SUCCESS in 7m 39s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/125db915ed7246819485d264110f6d64 : SUCCESS in 9m 19s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/02caac6fbabb46cab649bae7026a4f63 : SUCCESS in 33m 08s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cff2215f4a8478ca4fe29db6dc66b8e : SUCCESS in 24m 07s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/4376224a431d43c4af942381f2efd1f4 : SUCCESS in 36m 55s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96e719460b7048268d173683dc2fcfce : SUCCESS in 37m 24s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/8befd38b31ab4abb8f5fea688e3209eb : SUCCESS in 45m 29s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/f2757da95896466b84e8e7c2ef95a6f8 : SUCCESS in 37m 12s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/277bbff4dd0d454b82b5d6f50184336f : SUCCESS in 29m 52s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd8cad458a5d4eebbc274964aca5677c : SUCCESS in 33m 24s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/416ac0f9795e44a59a54ac3256a7b831 : SUCCESS in 39m 58s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfd7ba13c4624f0c9fb908e9fd2974da : SUCCESS in 54m 03s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/4d892fa255d94eb8bfb2f049df236222 : SUCCESS in 1h 10m 17s","accounts_in_message":[],"_revision_number":3},{"id":"394a2d318748f98453d81eae17fb86c2a64bb7ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 21:29:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bfd3b2fc8e4dd19cd8a1c488c1d1871bd5be446e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 21:34:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/543879b2aacb448d831754a2d7b5147c : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":3},{"id":"017b3bc3fdc2452ad08cea6446de4d095be0cf1b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-09 08:14:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9941e4439a7f04a09bfff761f40f6b1a299d0f5c","revisions":{"deeda9cd5bc54c2c22d66469dc6bde8c0b9955d9":{"kind":"REWORK","_number":1,"created":"2022-02-03 15:11:02.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/87/827687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/87/827687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/87/827687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/87/827687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/87/827687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/87/827687/1"}}},"commit":{"parents":[{"commit":"6a10fcd226b3edbaf8cf7a0db6ef374f844c19ff","subject":"CI: store bifrost.log as a Zuul artifact","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6a10fcd226b3edbaf8cf7a0db6ef374f844c19ff"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-03 15:10:23.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-03 15:10:23.000000000","tz":60},"subject":"CI: properly report failures in the upgrade job","message":"CI: properly report failures in the upgrade job\n\nChange-Id: I282f690214917eb4c17adfefc70a2ecddb7144fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/deeda9cd5bc54c2c22d66469dc6bde8c0b9955d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/deeda9cd5bc54c2c22d66469dc6bde8c0b9955d9"}]},"branch":"refs/heads/master"},"9dd1f0179317d54498fc515535fde9c82d6f4e9d":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2022-02-08 08:06:17.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/87/827687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/87/827687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/87/827687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/87/827687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/87/827687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/87/827687/2"}}},"commit":{"parents":[{"commit":"629bf5224864e032574f4a26e0e965bf1ba47782","subject":"Update /etc/keystone ownership on upgrade from Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/629bf5224864e032574f4a26e0e965bf1ba47782"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-03 15:10:23.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-08 08:05:55.000000000","tz":60},"subject":"CI: properly report failures in the upgrade job","message":"CI: properly report failures in the upgrade job\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/828178\nChange-Id: I282f690214917eb4c17adfefc70a2ecddb7144fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9dd1f0179317d54498fc515535fde9c82d6f4e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9dd1f0179317d54498fc515535fde9c82d6f4e9d"}]},"branch":"refs/heads/master"},"9941e4439a7f04a09bfff761f40f6b1a299d0f5c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-08 09:17:49.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/87/827687/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/87/827687/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/87/827687/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/87/827687/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/87/827687/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/87/827687/3"}}},"commit":{"parents":[{"commit":"72ee1ff418a992fb93f934c7de99f9bee570afc5","subject":"Only remove old services after they are stopped","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/72ee1ff418a992fb93f934c7de99f9bee570afc5"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-03 15:10:23.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-08 09:17:40.000000000","tz":60},"subject":"CI: properly report failures in the upgrade job","message":"CI: properly report failures in the upgrade job\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/828178\nChange-Id: I282f690214917eb4c17adfefc70a2ecddb7144fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9941e4439a7f04a09bfff761f40f6b1a299d0f5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/9941e4439a7f04a09bfff761f40f6b1a299d0f5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
