)]}'
{"id":"openstack%2Fbifrost~841142","triplet_id":"openstack%2Fbifrost~master~I66310701ad116411b20a2a02f032a23f5e674595","project":"openstack/bifrost","branch":"master","attention_set":{},"removed_from_attention_set":{"2218":{"account":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"last_update":"2024-09-13 20:39:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I66310701ad116411b20a2a02f032a23f5e674595","subject":"temp to use the snponly efi","status":"ABANDONED","created":"2022-05-09 16:14:09.000000000","updated":"2024-09-13 20:39:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6070ea3fe4db8cfdbc34bad3a065304579d870a5","_number":841142,"virtual_id_number":841142,"owner":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-09 17:10:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-09 17:10:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3652c61ae2f3a6ab96b720901f82729a651e75b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2022-05-09 16:14:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1feeff6e06a09984f736ae45e067f6b069e6ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 17:10:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c227e983eca4cd19845b0812de32db1 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e8711101293404d8f74727dd23e9cbe : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b8ca5548d1b4ddfb2e68486582c5d36 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/373dbc8fedf743a5a6b06c5fa0d3d0d6 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580e17a9b8474473aaa91a5fca06adc6 : SUCCESS in 4m 35s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/c447e9e7f1d4436a9c9a6c86a8b33334 : SUCCESS in 5m 46s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/70193d7897554d7ea4e70f076ee0569f : SUCCESS in 26m 38s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8b00a25fe1e490283118d0811fca961 : FAILURE in 7m 27s\n- bifrost-integration-tinyipa-centos-8 https://zuul.opendev.org/t/openstack/build/8509c1cfd06f4217af1851588baf94a0 : SUCCESS in 24m 39s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2c22edc661842e8abec94666df64add : FAILURE in 9m 52s\n- bifrost-integration-tinyipa-keystone-centos-8 https://zuul.opendev.org/t/openstack/build/0402414d4c4c44519c7c95741029bc34 : SUCCESS in 29m 59s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/8bc3376f39d94598a2b434bb9acfbe2e : SUCCESS in 28m 45s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/77523fdc97b74e6f80297e4d33d56453 : FAILURE in 11m 52s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32dd7b09f6794e27ba1291cb0efe0150 : FAILURE in 7m 48s\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/65f331b17cf945f3ac2aeb5b8a94635a : SUCCESS in 31m 41s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/0e09992d9c5a4b8a9aeac488c337ec9e : SUCCESS in 28m 33s\n- bifrost-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2fdcb053500409d8b6d09cea12ea049 : FAILURE in 25m 02s\n- bifrost-upgrade-keystone-centos8 https://zuul.opendev.org/t/openstack/build/c58225c491dd41088e73fd33e6202538 : FAILURE in 20m 37s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a5c57ae483c495aafa3142bf1dc8c39 : FAILURE in 7m 56s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/2435ad347a834dc38cde2e0c4ebf2857 : SUCCESS in 30m 03s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/dbf0c5e8c7fa4a93a912686770db6985 : SUCCESS in 30m 26s (non-voting)\n- bifrost-integration-tinyipa-debian-buster https://zuul.opendev.org/t/openstack/build/1c7bce5c24ef4fd4a59540b2efbeaac3 : FAILURE in 16m 38s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/5569f2053173410cb94d7a2d2eb77293 : SUCCESS in 25m 24s (non-voting)\n- bifrost-integration-dhcp-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f687bc79273c4a799949dbfe6e5841d3 : FAILURE in 9m 14s (non-voting)\n- bifrost-integration-dhcp-centos-8 https://zuul.opendev.org/t/openstack/build/d52fc0ea08ea4ab6b977f7abc6b44a50 : SUCCESS in 35m 20s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/7caf59371bd143229203ca5a07e0fe33 : SUCCESS in 32m 42s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9e9b2ccc1ab4ecaa2d05d2d5e1cebaa : FAILURE in 9m 02s (non-voting)\n- bifrost-integration-dibipa-debian-centos-8 https://zuul.opendev.org/t/openstack/build/bb22bfd0d2d441a28f29c3da544fa6e8 : SUCCESS in 33m 38s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/a18a96e017ca472d9b0dd491879163f1 : SUCCESS in 29m 41s (non-voting)\n- bifrost-cli-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b59f3470845b4713a5e8f72ecc576144 : FAILURE in 10m 14s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/128f17ae21e24b858455c5bffefed1b3 : FAILURE in 7m 07s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/bf56ee23d7324110bad7b52737040b79 : FAILURE in 11m 30s (non-voting)\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/fb6d488f0449457bb44b3deb216fe036 : SUCCESS in 30m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6070ea3fe4db8cfdbc34bad3a065304579d870a5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2024-09-13 20:39:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"351d1e7e49741e53ebc6fefe90d56d1a2a235f49","revisions":{"351d1e7e49741e53ebc6fefe90d56d1a2a235f49":{"kind":"REWORK","_number":1,"created":"2022-05-09 16:14:09.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/42/841142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/42/841142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/42/841142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/42/841142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/42/841142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/42/841142/1"}}},"commit":{"parents":[{"commit":"7640a231a84f87939d3242b062f4ca1e89bcfa4a","subject":"Allow for almalinux and rocky installations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7640a231a84f87939d3242b062f4ca1e89bcfa4a"}]}],"author":{"name":"Adam Young","email":"adam@younglogic.com","date":"2022-05-09 16:13:59.000000000","tz":-240},"committer":{"name":"Adam Young","email":"adam@younglogic.com","date":"2022-05-09 16:13:59.000000000","tz":-240},"subject":"temp to use the snponly efi","message":"temp to use the snponly efi\n\nChange-Id: I66310701ad116411b20a2a02f032a23f5e674595\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/351d1e7e49741e53ebc6fefe90d56d1a2a235f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/351d1e7e49741e53ebc6fefe90d56d1a2a235f49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
