)]}'
{"id":"openstack%2Fbifrost~881309","triplet_id":"openstack%2Fbifrost~master~I0b92abde82e73379058880c658727bcf5f0a8096","project":"openstack/bifrost","branch":"master","topic":"ansible-lint","attention_set":{},"removed_from_attention_set":{"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2024-03-22 02:11:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0b92abde82e73379058880c658727bcf5f0a8096","subject":"ansible-lint enforce meta-no-tags","status":"ABANDONED","created":"2023-04-24 05:12:08.000000000","updated":"2024-03-22 02:11:02.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5af67491b5755983f69d7efeba30ac084e4e8c97","_number":881309,"virtual_id_number":881309,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-25 23:15:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-04-25 22:06:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2023-04-25 22:06:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}],"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-24 06:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-24 06:55:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-24 16:53:21.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"CC"}],"messages":[{"id":"0487ccb86c7771c88a09e5b603b5523bd6746973","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-04-24 05:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e016fe1c0567788c3f4b281aa3585969c8d59ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-04-24 05:31:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"068e904997d8f8ecb44bc3aa37dc7e9edff74696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 06:41:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f58a71e89654b16937b78ba95cecc2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa2431fce5e3493baa2e1ba7fb52b75e : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/493a6c16e28e415a906d610373fdeb1f : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd697b57d51a41278ca0b285f43ebd10 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b59ed8aa3d46238329f85d838e41c0 : SUCCESS in 5m 03s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/92f7e1060fe74cc6b174d04e912e2555 : SUCCESS in 6m 25s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/a27c86220ecb46ac991512893389a992 : FAILURE in 40m 42s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c73a48de536e4c25affc25c1a4b125c1 : SUCCESS in 30m 56s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bcc2f3a612848cf90b989d405f364bd : SUCCESS in 39m 54s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/91a4227d10d7445291861ee2311d37db : FAILURE in 44m 04s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/00917f55979f45b6ba832b082d671853 : SUCCESS in 43m 03s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fa19a23aefb4924b55df0ad0933e482 : SUCCESS in 35m 02s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/cc2c1e66fc7949e49b010f1884e0c957 : FAILURE in 42m 46s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43f0af075fe54f468f417de3023bc339 : SUCCESS in 1h 04m 12s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/a6685bf1eadb4062b625eb402446bdfa : FAILURE in 43m 43s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79d28c7c35c54c489bb551e832777fbc : FAILURE in 10m 50s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/cfed3448e9704abbbc67526139f73fa4 : SUCCESS in 47m 33s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/d55ba4809a734c1a84512384487fa0bb : SUCCESS in 38m 03s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b9820183b804f279b634e5e66624a19 : FAILURE in 39m 34s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/9f251ca6c0eb4fa7ac2c93ed9cd70403 : FAILURE in 42m 54s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b84d6437de8f4c0e8bff21b99d47cfa8 : SUCCESS in 27m 42s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/85d50d7dac8945bc82264de1f0d97b1d : FAILURE in 47m 17s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06d5de5e93f74fd794b7639517e2327c : SUCCESS in 32m 43s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/93412726e05f4235b05da2f011ce99ed : FAILURE in 8m 40s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/7ee26b93bdba43b09d9bf722ab0b2af9 : FAILURE in 9m 48s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/0a31a274b6674fe9bc22b6afedd816f9 : SUCCESS in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba767843703f0c60fcf4e9f46021042663ef33ca","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-24 06:55:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cb50c50315a6e8c1b91c1d113b346da200336ec9","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-04-24 16:53:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a54b1d3501cc250b25c89f93d4688469e642980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-04-25 22:06:00.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bb4237dacf9537b058ea4198811ed77ace27ca72","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-04-25 22:06:51.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8e29fe5fa6bbf0ef6a7805f222d25b12db996343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 23:15:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74a7f121e19a46c39e968b1c5bb4a074\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66d7c6ed070b460d99a5a8c17079eb20 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c19db7680cd4b108c758bf8384c3ff0 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/775be1616d8d4c9da0fee22486c37d9d : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dae87c7eb214257b2eca8b788111c1a : SUCCESS in 4m 54s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/351b38365d5d4a1eb156aa8cde7bb96b : SUCCESS in 5m 37s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/1f4d2c303f2f4e0ea3493faaa024151b : FAILURE in 40m 30s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/918ebecd81cf4ee0ad0c06b4858cfe11 : SUCCESS in 41m 31s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77bb12fa4edf4cb9b95645dad1944803 : SUCCESS in 39m 56s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/5f087b0e3dc847a1971729820ae7e018 : FAILURE in 45m 17s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/a83b1e53891f4dd7a4afb19abe6a3879 : SUCCESS in 29m 04s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08f60c41164d450ca7b35e47acf02bcf : SUCCESS in 42m 14s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/317a95a4f39b4c20b1c3e1017abb8084 : FAILURE in 42m 38s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2c9edf82d2c4a4d9b2bd6827342d746 : SUCCESS in 1h 04m 53s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/d578fc68645a41b6b38a77c79ea64625 : FAILURE in 46m 21s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80f2a994130f4e3a84506cc730eb6bee : SUCCESS in 41m 45s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/094261e63e2e4c6c987bf2bc160f5e3c : SUCCESS in 45m 41s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/e92195a3704b41af8516182f73cc8698 : SUCCESS in 41m 52s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0363149d5df847309eb312443338b519 : FAILURE in 40m 37s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/7eb7c05098d34492a9f582d1bee798ed : FAILURE in 42m 46s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38b49aebf42746c3a24ef86a71658be9 : SUCCESS in 31m 06s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/a8792e0253c14257973a716bcdcc6690 : FAILURE in 49m 04s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c39d10b8bad746388739a83f52bd182e : SUCCESS in 37m 05s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/5ba9f3b10aa24b668d129027d81d3c6e : SUCCESS in 45m 51s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/e3ba425630034107a268fdbb55a2cd42 : SUCCESS in 39m 32s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/bf2b56da18f1446fbab8777c33334186 : SUCCESS in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5af67491b5755983f69d7efeba30ac084e4e8c97","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2024-03-22 02:11:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ebb9616df0e1f54f919be459c882be308c1a62f","revisions":{"7e99fc2de491d2bbff9c4f24bac1fc7ca35f53f8":{"kind":"REWORK","_number":1,"created":"2023-04-24 05:12:08.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/09/881309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/09/881309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/881309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/881309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/881309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/09/881309/1"}}},"commit":{"parents":[{"commit":"f3637ffc467195fe64ca04975dfb091255eae0da","subject":"ansible-lint enforce key-order[task]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f3637ffc467195fe64ca04975dfb091255eae0da"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 02:56:44.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 03:00:53.000000000","tz":720},"subject":"ansible-lint enforce meta-no-tags","message":"ansible-lint enforce meta-no-tags\n\nmeta-no-info is also now enforcing, which required no changes\n\nChange-Id: I0b92abde82e73379058880c658727bcf5f0a8096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7e99fc2de491d2bbff9c4f24bac1fc7ca35f53f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7e99fc2de491d2bbff9c4f24bac1fc7ca35f53f8"}]},"branch":"refs/heads/master"},"d5654eb4b05496cb6b594c82227f993d27823894":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-24 05:31:23.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/09/881309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/09/881309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/881309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/881309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/881309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/09/881309/2"}}},"commit":{"parents":[{"commit":"2d02ee86338c811eaf531601bb5efadef29148b8","subject":"ansible-lint enforce key-order[task]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2d02ee86338c811eaf531601bb5efadef29148b8"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 02:56:44.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 05:28:53.000000000","tz":720},"subject":"ansible-lint enforce meta-no-tags","message":"ansible-lint enforce meta-no-tags\n\nmeta-no-info is also now enforcing, which required no changes\n\nChange-Id: I0b92abde82e73379058880c658727bcf5f0a8096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d5654eb4b05496cb6b594c82227f993d27823894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d5654eb4b05496cb6b594c82227f993d27823894"}]},"branch":"refs/heads/master"},"6ebb9616df0e1f54f919be459c882be308c1a62f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-25 22:06:00.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/09/881309/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/09/881309/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/881309/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/881309/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/09/881309/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/09/881309/3"}}},"commit":{"parents":[{"commit":"d75ac1ee36dde59901edb8bd8ddc4215b8ab477a","subject":"ansible-lint suppress jinja[spacing]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d75ac1ee36dde59901edb8bd8ddc4215b8ab477a"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 02:56:44.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-25 22:05:01.000000000","tz":720},"subject":"ansible-lint enforce meta-no-tags","message":"ansible-lint enforce meta-no-tags\n\nmeta-no-info is also now enforcing, which required no changes\n\nChange-Id: I0b92abde82e73379058880c658727bcf5f0a8096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6ebb9616df0e1f54f919be459c882be308c1a62f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6ebb9616df0e1f54f919be459c882be308c1a62f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
