)]}'
{"id":"openstack%2Fbifrost~881330","triplet_id":"openstack%2Fbifrost~master~Ie30a1278c52857cd8f57aa442d62af93f8aca0b0","project":"openstack/bifrost","branch":"master","topic":"ansible-lint","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-22 02:10:54.000000000","reason":"Change was abandoned"},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2024-03-22 02:10:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie30a1278c52857cd8f57aa442d62af93f8aca0b0","subject":"ansible-lint suppress jinja[spacing]","status":"ABANDONED","created":"2023-04-24 05:12:08.000000000","updated":"2024-03-22 02:10:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"876a2292460cd293f08e331ee47b78ee0f7d3239","_number":881330,"virtual_id_number":881330,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-25 23:21:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-24 06:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-24 06:54:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"67ef8122a56ce71f9a61ecfffd3a1977952554fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-04-24 05:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5d2cb50fe3b5a348c681cd480e61266c2f76763","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-04-24 05:31:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2a186157c82c4e07bb101c1f9539ddbfd852c456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 06:40:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5863d15a6af84f37a5236bf4d206ca3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5cc9befc2f4e0fb647f7c2af9d29e9 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a78a2fd994047d2929c31ec01d12ec0 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa9ff233a2c44be38268bf089cc3f40c : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6cdc7785154579b1594a2ceca67063 : SUCCESS in 4m 50s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/35b24f22210044daa954c8185cd5b568 : SUCCESS in 4m 44s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/ff32fd8362324cf6a4a00f9810be65d6 : FAILURE in 42m 00s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/136c8807e9ca4a8c9c29cd75b65a9642 : SUCCESS in 38m 24s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4009d516e08440408d244164d90197c3 : SUCCESS in 36m 23s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/0d1d43b5aa9c44f39896600d326da00f : FAILURE in 45m 21s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/078b18db7e8644caa1133b2885130c9a : SUCCESS in 37m 35s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a823355ef5a74b36927877b1b4f27b8e : SUCCESS in 34m 39s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/69bb72bab8874f53ad34a9e153545a8f : FAILURE in 40m 24s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ee45a9a55914b3c80b8b83aa6a35f97 : SUCCESS in 1h 03m 45s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/99ddbf1b34074e94a37301a2b47eb74c : FAILURE in 43m 24s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f51895a77d84c0f954f92310db3e25b : FAILURE in 11m 59s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/2be69cc8ba4d47ec9b20ea5a05dbe7db : SUCCESS in 45m 46s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/674ec9d81368419b87d6838da1924159 : SUCCESS in 46m 02s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e5202392573447aa70846b09eada5ba : FAILURE in 39m 00s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/cf1793e9d6b64a72b73c2293046c5859 : FAILURE in 40m 36s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0c1484fd0b94b59b18f622dcc564918 : SUCCESS in 25m 54s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/6cd32edf5cf8416d979c77d82e517d44 : FAILURE in 51m 37s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca83cacc4a5244fa854c70d5eaaab73b : SUCCESS in 40m 33s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/ba7d82d8d7744d059e3b35b25ed20fe0 : FAILURE in 8m 34s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/02da35401c0442fb9d04aec008f34224 : FAILURE in 9m 07s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/4a13b3e6780c48989a9571b6e4bffe26 : SUCCESS in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71e63ad274849bdc1f4dd29a8d43c6a68fd61cf5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-24 06:54:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"96cd561976ddbb91c1ecbe6277ebe67dc5727fad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-04-25 22:06:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1a4d46ebdd30b63d95ac897309f7195f4f3d78d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 23:21:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33eddec85f0f45909fb62bb03fd79051\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159f750995614de88e887eb7881d5283 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a98786a0a57466980626feeb04ebf5e : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f2f65f62fdf47088f29e6d29118ddf2 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3f7aad7af54874b0a25b9e99c43215 : SUCCESS in 5m 31s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/1a56890400a8451eab2489b7af295281 : FAILURE in 4m 20s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/45ea7599789b492ab51dd14376f9971d : FAILURE in 43m 21s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56a1745efa3b407ebc51f60114842368 : SUCCESS in 31m 34s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bee96324a0824af89b91025fe1df09fd : SUCCESS in 37m 51s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/72cc7186e6994f0bb0012fbdf47d041b : FAILURE in 46m 58s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/dbba501e39314847a3386126758cd457 : SUCCESS in 38m 46s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9292c354b9534f1d85b4d529852ed540 : SUCCESS in 40m 08s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/6a4cccb41f4243b89755c35b687f2f06 : FAILURE in 42m 07s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddff958f2c4946e0a2a1481cdc6602e9 : SUCCESS in 1h 09m 48s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/d099081e65794e43a3811e0f0418df3d : FAILURE in 43m 33s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24b7d6a9ba5f47b89a647bd607d4b514 : SUCCESS in 37m 56s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/15a8e8b7341343429a5275e5d7315432 : SUCCESS in 51m 56s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/dd7bd89ea32e4d01847010f83f95b8a4 : SUCCESS in 41m 52s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a0d438e1e11478a91912c9327062231 : FAILURE in 38m 34s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/c57a50e4b45e410b94fd37003bbef236 : FAILURE in 42m 13s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/704e9ff22e0248d791d2d88b329c40db : SUCCESS in 35m 01s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/d901754999d64180b73d239c2d123c7b : FAILURE in 49m 39s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1fb0774ac2b4f479037ff7fc633a602 : SUCCESS in 37m 10s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/4a02c7d4c5b4492d9ecf0397bcc44d49 : SUCCESS in 37m 58s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/ad3bbf0327544a91b75b7513f3742ef2 : SUCCESS in 32m 48s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/6ae9d09d1b1a4217a5ea053051133248 : SUCCESS in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"876a2292460cd293f08e331ee47b78ee0f7d3239","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2024-03-22 02:10:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d75ac1ee36dde59901edb8bd8ddc4215b8ab477a","revisions":{"867edd1f494ce2d8e8deac7c382c3383d1a0041a":{"kind":"REWORK","_number":1,"created":"2023-04-24 05:12:08.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/30/881330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/30/881330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/30/881330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/30/881330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/30/881330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/30/881330/1"}}},"commit":{"parents":[{"commit":"7e99fc2de491d2bbff9c4f24bac1fc7ca35f53f8","subject":"ansible-lint enforce meta-no-tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7e99fc2de491d2bbff9c4f24bac1fc7ca35f53f8"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 03:04:11.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 03:04:11.000000000","tz":720},"subject":"ansible-lint suppress jinja[spacing]","message":"ansible-lint suppress jinja[spacing]\n\nThis rule is causing warnings, and in many cases would result in less\nmaintainable mult-line jinja expressions. Disabling for now.\n\nChange-Id: Ie30a1278c52857cd8f57aa442d62af93f8aca0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/867edd1f494ce2d8e8deac7c382c3383d1a0041a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/867edd1f494ce2d8e8deac7c382c3383d1a0041a"}]},"branch":"refs/heads/master"},"0d512088764a8f60f76b039d45696354786dc18d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-24 05:31:23.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/30/881330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/30/881330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/30/881330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/30/881330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/30/881330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/30/881330/2"}}},"commit":{"parents":[{"commit":"d5654eb4b05496cb6b594c82227f993d27823894","subject":"ansible-lint enforce meta-no-tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d5654eb4b05496cb6b594c82227f993d27823894"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 03:04:11.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 05:28:53.000000000","tz":720},"subject":"ansible-lint suppress jinja[spacing]","message":"ansible-lint suppress jinja[spacing]\n\nThis rule is causing warnings, and in many cases would result in less\nmaintainable mult-line jinja expressions. Disabling for now.\n\nChange-Id: Ie30a1278c52857cd8f57aa442d62af93f8aca0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0d512088764a8f60f76b039d45696354786dc18d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0d512088764a8f60f76b039d45696354786dc18d"}]},"branch":"refs/heads/master"},"d75ac1ee36dde59901edb8bd8ddc4215b8ab477a":{"kind":"REWORK","_number":3,"created":"2023-04-25 22:06:00.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/30/881330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/30/881330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/30/881330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/30/881330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/30/881330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/30/881330/3"}}},"commit":{"parents":[{"commit":"37f1bfe83f22e8f5569a2340a26e816fd3a9917a","subject":"ansible-lint enforce key-order[task]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/37f1bfe83f22e8f5569a2340a26e816fd3a9917a"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-24 03:04:11.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-04-25 22:05:01.000000000","tz":720},"subject":"ansible-lint suppress jinja[spacing]","message":"ansible-lint suppress jinja[spacing]\n\nThis rule is causing warnings, and in many cases would result in less\nmaintainable mult-line jinja expressions. Disabling for now.\n\nChange-Id: Ie30a1278c52857cd8f57aa442d62af93f8aca0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d75ac1ee36dde59901edb8bd8ddc4215b8ab477a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d75ac1ee36dde59901edb8bd8ddc4215b8ab477a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
