)]}'
{"id":"openstack%2Fbifrost~883943","triplet_id":"openstack%2Fbifrost~master~I8b3a973bd5ef19333af563c170b2a28fbbfb5c76","project":"openstack/bifrost","branch":"master","topic":"generalise-use_cirros","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-09-05 17:05:38.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-24 10:48:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I8b3a973bd5ef19333af563c170b2a28fbbfb5c76","subject":"Replace Cirros references with generic descriptors","status":"MERGED","created":"2023-05-23 14:25:35.000000000","updated":"2023-09-05 17:07:45.000000000","submitted":"2023-09-05 17:05:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883943-generalise-use_cirros","meta_rev_id":"98353fb94ce143da8f95138e786f07760557e37a","_number":883943,"virtual_id_number":883943,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-05 17:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-09-05 15:45:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2023-08-24 12:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2023-09-05 15:45:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-23 14:25:35.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2023-05-23 16:09:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-08 10:36:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-09-05 15:45:09.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"e308fe8f8017bdf3b2ff3658716ee5bd2c332c21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-05-23 14:25:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3386956fe737825cf60d3eae45a0c5257e408648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 16:09:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c36f5121ff744e8b9650b24d14082cf4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4d541dcee240daa9e774e83fe5ffc6 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/937324081e9448d39e83f6b75b91481b : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a65d588e95e041f0b2814ffbdb89d180 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5279b8ce6e04ee297cac65165e7e2f6 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b87d44bf0f48869d6bb76307a6d644 : SUCCESS in 6m 27s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/9292add5f1af4a059336b43c435d271b : SUCCESS in 6m 19s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/68c5b07c8c5a4d83b4027150e520f301 : SUCCESS in 44m 58s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48734a04f41a445a8569c9e73c4b4895 : SUCCESS in 30m 09s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd1137b557c14b82b9b76e95ddf8efa7 : SUCCESS in 35m 19s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/d64a828de0a149e7a67d8e1540e5ba4f : SUCCESS in 48m 28s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/cae0a89f3ba3496a9a5a61f075f1b96a : SUCCESS in 38m 42s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b58b68daba2143d2833a121e5b54a130 : SUCCESS in 32m 57s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/9bad6d217e404f73a05b6ab8d6ffc5a9 : SUCCESS in 34m 28s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f40593aed4da48e3be8be76597ba6698 : SUCCESS in 1h 03m 50s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/e52ae9d547f145fd91e235656fb78837 : SUCCESS in 1h 21m 13s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e54d68dce4d54f2cad6a20ead4ee867b : SUCCESS in 45m 45s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/97cde96cef1b4901ac91164ea0bb4bb6 : TIMED_OUT in 1h 00m 48s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/74aaf1527f28478ea7c9d3f27e311c2d : SUCCESS in 47m 42s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f4ebc2bce414f5eb1152703f086ea80 : FAILURE in 39m 28s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/b6ae80666d304c3995469035e8ad6f67 : FAILURE in 42m 26s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95d2032c54344381829b13842db99c39 : FAILURE in 47m 10s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/15c6d1a603d442bebc6ec791712190b8 : FAILURE in 54m 11s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2219a66a7a374e0db955dd76eea8d6da : SUCCESS in 34m 02s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/5b661a8801934dd69e9c1af058e109b1 : SUCCESS in 37m 43s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/d4693589b84a40e683595023951d83ee : SUCCESS in 42m 51s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/abd34f0e493947838768a0ebe8826349 : SUCCESS in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc4ee801adb43a8fc2da6fdca745f7d602141e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-05-30 09:14:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b0ecbf6783c9c40499bf1ef9bfbff11e7f4d6af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 10:31:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70ff34eeafb747e2a37599587551917d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cdc5efe9adb46b6846105a67d2f46db : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d602e7efa86d4c51a35ac959fbe69047 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65c06b2fb5f64ec2b2568631b03b4325 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95073c9b83b647e2992b4853e84ff69d : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2ed7712d464a87957ebc7a31a2d869 : SUCCESS in 5m 23s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/2b2a41f888664986a1c79966fb077880 : SUCCESS in 7m 53s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/5781cf91fc6845de980c89bf618fb09f : SUCCESS in 47m 48s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/248faed110524b2d85f0616863014047 : SUCCESS in 40m 44s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f2b3674c04b4d7ca515656b409525ce : SUCCESS in 34m 51s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/22c70d15d12f44708beaadad83da3d73 : SUCCESS in 54m 11s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/11e0e1fe6f5448208b89f81efb35e524 : SUCCESS in 33m 46s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd08d02224074946ac3ad58307b613be : SUCCESS in 47m 08s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/08a9a5ce54444fbc93c031129c5cd00b : SUCCESS in 39m 09s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57ee0e4f3c81431892b8ba4a75780e49 : SUCCESS in 1h 09m 39s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/09518c1e400f47e9ab9b3dfb48a23e97 : FAILURE in 1h 08m 48s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ded14eca58134337a852458fca68dbdd : SUCCESS in 40m 40s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/a4f5ba6ba657460e9a40a3bbbf7f7772 : SUCCESS in 53m 39s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/01259405ce034350901678f5559ef859 : SUCCESS in 43m 23s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1629b1fe52e14fbab1bd4aed013c6e00 : FAILURE in 39m 46s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/8307b3ca6ec74f439d86fc08304c1c54 : FAILURE in 42m 45s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9aa86dd00d1d4d29b9ecd0f0451bc6ed : FAILURE in 9m 18s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/24408716c97c422d9f5d7d8663ab7186 : FAILURE in 12m 15s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc085afbaadb49459545755ca4b302a9 : SUCCESS in 39m 56s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/313d3b0b7d2d4805abbf833e0ee4ced9 : SUCCESS in 41m 57s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/6b14e303f71a4eb3be7a26ff7dc7812a : SUCCESS in 45m 05s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/9fd8efc395e8471fb4abb47e9eded61c : SUCCESS in 26m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7e1207f8784e721536f41fbb7c199a25762dbf4","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-05-30 11:23:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2fcb45cf300ba35282dd9cfb2b821a6a6830d806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:37:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8ae6bec7c4946c78fae25b36cf7c112\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ffa6c81085749d49a4c792dc541c497 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3cd90c2adbe4575be1a72b48e11e4c3 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff7cfbda649c44ceac1ca95e16f5a2a4 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c106b4f7e7c48f7a35d7885ced250eb : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c2008b2429a445abf3b5144b42e3241 : SUCCESS in 5m 34s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/b79ea29197d44aeeb9f42765ca6883ce : SUCCESS in 6m 53s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/804c390a62f846509581a5f8049428ba : SUCCESS in 39m 24s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/688f7251839944be979c5a248fb1c895 : SUCCESS in 31m 12s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be44896888e24245a85f0a807adab670 : SUCCESS in 35m 11s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/52ce4f72556945d187781fe6ceff31c4 : SUCCESS in 54m 02s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/a8cd2d8769e24993a73dda802f4a452a : SUCCESS in 38m 53s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b9821df99444f96ba3dd109426892a3 : SUCCESS in 36m 10s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/6da88d48ebf8467284414a0bf8482d02 : SUCCESS in 30m 54s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/765f32a1eff24b85afa76452fb546275 : SUCCESS in 56m 13s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/f11f52796d1c43ffb307f9a1f3ff5833 : SUCCESS in 1h 12m 32s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bbb1aabd8e446ffad7709ebf3c71553 : SUCCESS in 29m 45s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/daaf07c559404dac9a6441922445dbe6 : SUCCESS in 57m 18s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/f126286a9b0846579eb74a79c96a82cf : SUCCESS in 45m 12s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6152af7cc1d54f68b158662013dbd2c4 : FAILURE in 38m 23s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/b88846b79cf7466e966c0dac936f7b97 : FAILURE in 42m 43s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/338ab2b6c8084af6858e069fb70f85f3 : FAILURE in 7m 36s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/90294e7861b44fa488635f569af83653 : FAILURE in 11m 19s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12525e9e8f344aacbc4bc18cad216b44 : SUCCESS in 44m 25s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/007c4d35c4a64b65ad59905c9f97042a : SUCCESS in 39m 01s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/81d409bd9d344f4c83fba78148ce1b56 : SUCCESS in 40m 24s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/e9af4338d6cd4880ac7cbcb660de4ef7 : SUCCESS in 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"970be0cd81ca1ed6e49cb99cd95b1c4ee6f992f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-05-31 13:41:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"43c27a16cdf2fb99ee4714deea6c5225b4744bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 15:12:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2121dbe4819449598554cb5c1306134d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489bcc1cf8364ebebf372fc240200cb2 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57b55dc666664a83b1a10df740e88bee : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b143583a85443cc8e3638b25541d546 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11d07df557d04107ae85f4129ee6b476 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d195907678e04017b885873790a8ed34 : SUCCESS in 5m 30s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/e287fd2ba67d462daeb1f23f7ec3d387 : SUCCESS in 7m 18s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/08ca775dc6574aab94df5226e2267d68 : SUCCESS in 50m 31s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1abe3d8341f43d88f3185c80a97a8d9 : SUCCESS in 37m 43s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca2115c3f073468492e17ecd23caed9f : SUCCESS in 34m 47s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/c6800c50aff945c5b5caf94b2c034948 : SUCCESS in 46m 34s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/8508f760e2df4a04b1e5d8df07c8a5c9 : SUCCESS in 34m 52s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92a26c4d63c84345acf7e553b47ecc9e : SUCCESS in 33m 00s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/030d4849e4c8410d89521ab66aca760f : SUCCESS in 37m 53s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a58cf5af2b84e88989209aef37774cd : SUCCESS in 1h 00m 00s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/51cbe24c3dad4d85971139be6bdd9d4f : SUCCESS in 1h 23m 11s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b90fc1b7b0540ffb1d8435256ac2a2b : SUCCESS in 44m 51s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/3907d55b6b57483ca93ae1efc498e172 : SUCCESS in 55m 51s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/b2affa5c3acf400286f001f558d15e53 : SUCCESS in 40m 48s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73a53da08ada447dbea23c5fd0bfc6b4 : FAILURE in 40m 22s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/3b87169174fc44b8b7e87c7a89ba906c : FAILURE in 42m 17s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5e2a909c8be4a5b92d7c095f1f4227e : FAILURE in 9m 38s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/cccdf946578b44389161dfb7c5162e09 : FAILURE in 11m 43s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bd507b2fe33413d99e04933e8fde884 : SUCCESS in 41m 41s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/51ea4053ee4f46aeb3d663ce54161713 : SUCCESS in 45m 13s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/c6b85fba00e44e3086a2d5906cd04bfa : SUCCESS in 40m 00s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/d6a515af84c64b3ea76b531eb86035bc : SUCCESS in 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8549a66a4535825f89f8856405acd0d1d92dd150","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-08 10:36:52.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6c4bac7817b8e8b92e2b2ccca885e2ca0851e8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-06-21 15:02:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"789617da4ac92ce8653c650ce73e0c5b643b6e34","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-06-21 15:04:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d7906458e3a0a0035c36e79843f329de90f0ca55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 15:59:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b66c919e8cfd471a8e04889a6a7f59f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0edac4e64b647e69b981762dd84c4a2 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2aac59bd7ef74f73905a4590c317d953 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8705321ad9e5410f9ad99fd564ba83a8 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e9d633003474672bb5bc5494927fc8f : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8cfdc0fe804feebfd413f9d0783f13 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c53c287e5028433c8ed2a7eb86489f09 : SUCCESS in 3m 09s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/0b65a2e14c5640ff8794abf6ec20d544 : SUCCESS in 8m 07s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/915250c15d1143e39dd7f357e60a3754 : FAILURE in 9m 57s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ed3b3298fdc4c2ba2ad7671e0ea787e : FAILURE in 8m 47s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd8014296f124cc5a830325c808f7d64 : FAILURE in 11m 13s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/555ef547d49140f1bdb5dd1db4303f1c : FAILURE in 16m 51s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/141c9ad0928f449dadbdf3607cae0ca3 : FAILURE in 11m 14s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/268f6abbf7ff490fa055f2ccc63ef515 : FAILURE in 8m 45s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/ad85a8593e9142eca451763ac4e4e912 : FAILURE in 12m 23s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27d7a46781f94db3a9be221e19f5e11e : FAILURE in 10m 10s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/774526e4ea4242b7b4cd8ac8c4087a9c : FAILURE in 16m 09s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf7c9d9efbc94029b748ecbe25fa80a0 : FAILURE in 15m 43s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/67aeb9cc65be493bae85d7b9d202d639 : FAILURE in 13m 16s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/af4c107e3e204adcb0c9d35b8d523561 : FAILURE in 12m 51s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4cf5f6f689f4b1889ec4b7b24076c02 : FAILURE in 8m 13s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/206a2398195e4b8e85ca3d486443a026 : FAILURE in 12m 49s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61aa9f0c4a3e497482c940a0f3e08517 : FAILURE in 9m 05s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/23ccae7fa002466b9292dcf471ce5b20 : FAILURE in 12m 46s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/893efefdb5174524b0abbf6e741bacbb : SUCCESS in 44m 22s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/24afa13629c64e3eafe57199e45669d2 : FAILURE in 10m 06s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/f1f3b29a77874c5d9d1b12861b0072ed : FAILURE in 11m 32s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/afc68963c4fa4b709236663a3233d4c9 : SUCCESS in 25m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fc5e98ba6c91e49e314db4a53358b610727cc289","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-01 12:58:57.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d9d0e70d7a30c0e551cd12ae60372aec65a95f49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-08-09 16:12:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"47f6044033b4d0b8c5ca48e86d5f0be0c46ba548","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-08-09 16:13:31.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1fb48cb0f807a86584dd60e582c4f116830c9423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 19:00:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/589a058e9ee84a1eba3cec4d64620523\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be6245ac86741769da80fb6e9e8c593 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdca0908707f4dcf8abbc04483121015 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/435d6a3f0ebe433aa896afd0f88eb5cc : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c9160f675a64cd387c0644e83b0d9d9 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f13d3c3c544275ba7a291913bf9474 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cafd6a7b5ca420d84bc4a015adf9ae8 : SUCCESS in 3m 15s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/552213fa9c0949f0ac264a26e6d4a627 : SUCCESS in 8m 26s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/d6c41c93aa77419ea0f4433fd446d53b : TIMED_OUT in 1h 00m 47s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76fcfa64025f45019ae7c1736332934b : SUCCESS in 37m 06s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9063ca4017ae4a6a9390574756c8a321 : SUCCESS in 32m 54s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/ca93cf11117c42a8af61e7eee163c517 : TIMED_OUT in 1h 01m 02s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/a6bd837ad6a047fd97e23b9ed11ee704 : SUCCESS in 31m 49s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18fd2134dc4841d9a21359b37f0f8060 : SUCCESS in 38m 24s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/d9099548939941bbb2d5a484eafdae44 : TIMED_OUT in 1h 01m 19s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f961d8de9c8b4da39f4666f6532298fd : SUCCESS in 1h 02m 12s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/0ee79ed68408453da973af3ae1521233 : TIMED_OUT in 2h 41m 02s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a977013331024780b11ff3a9cd102f8c : SUCCESS in 35m 46s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/872b493130254c7ba989fff0714f046b : RETRY_LIMIT in 2m 08s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/811e718d6e1d4ffdb6d605e38a02662a : RETRY_LIMIT in 2m 15s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41c54cd332dc4c45a46e47382f04e3d9 : FAILURE in 39m 15s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/67010685978a4c42816423e7253d1c90 : TIMED_OUT in 2h 01m 03s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b3941e162db44c48a8297403dd33686 : FAILURE in 45m 07s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/6e26fe2095074e1a86346801aca31476 : TIMED_OUT in 2h 00m 59s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16e99f27aeb345e084fa3be2137d05f1 : SUCCESS in 37m 43s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/146265753af541768604aa74faa43fe6 : SUCCESS in 40m 31s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/079a59f72d8e4cacba9c83bf6461ad01 : SUCCESS in 42m 17s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/8f008ba69f4b441f8b6656bb95e4d53c : SUCCESS in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1c55d128c5a53f288e1710faa5085d7debac5c16","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-10 10:03:50.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"35ae7afdce8a304165769f2e0e51995357b61547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-08-10 10:41:39.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1135e62c3aef07b952e4789abc1c303452a58037","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-08-10 10:41:56.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"753960cabb3be8c0e758ca11c8717f41e609245d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 13:30:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cfe8abc0a214f0ea61ffcab874e8751\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd08f36066744f0f858486b99e3866b5 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10a155bbb56e46b69c1389c1ce2f0c13 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9a075bbda2b4a65b35bbb2713b6399f : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/797c729204a3444cac7109e339a587d9 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84343cc7daf045c1bb29448d58b4f55e : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e898922570e413b8a6e86c3a0c05f5d : SUCCESS in 3m 04s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/b91ffe70a7a740bdb1fab8b17d8f2ca7 : SUCCESS in 7m 34s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/bd26c92527c04693a7b067dc7576d344 : TIMED_OUT in 1h 00m 46s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/664645fa38754060a99f17a4a0d4c1e7 : SUCCESS in 40m 29s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2f3a2bcb9d14f3ab37a4ff0ee214c70 : SUCCESS in 47m 16s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/19b0403a8b834ab995fff01a62ba5484 : TIMED_OUT in 1h 01m 09s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/dfc7a26d47da4ddf9b26b11edb7cdf41 : SUCCESS in 40m 28s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/015b9509f1174e97835c22ce9e5391aa : SUCCESS in 30m 57s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/843238a821ed4fa29ab5871737580e65 : TIMED_OUT in 1h 00m 45s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4be59923f0f0478e94113de25b6bca0b : SUCCESS in 58m 34s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/1c499702b5274950a52bf50fd6a79fb5 : TIMED_OUT in 2h 40m 57s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7089b61a4784a409ba287a858858229 : SUCCESS in 38m 42s (non-voting)\n- bifrost-integration-tinyipa-fedora-latest https://zuul.opendev.org/t/openstack/build/b9b0b938030b418eabb9cfbd9aba4f90 : RETRY_LIMIT in 2m 19s (non-voting)\n- bifrost-integration-redfish-uefi-fedora-latest https://zuul.opendev.org/t/openstack/build/31d3b62b60be4fcea97db19ab6c75c3a : RETRY_LIMIT in 2m 03s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee3ada2e4cf943ec9342adaaf03ba904 : FAILURE in 40m 25s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/5fed35a7ca3a4c8e8eccb8849aa28b8e : TIMED_OUT in 2h 01m 07s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d03c9afe967342e084cc5fd2c271fe11 : FAILURE in 47m 33s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/3c7259307d44430786e145807c21fb68 : TIMED_OUT in 2h 00m 47s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27d164470e414fbea4f2044d227ba095 : SUCCESS in 32m 07s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/b30cca9f725543e184144b2a17574d25 : SUCCESS in 34m 56s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/6170da8b6120485e8faea354ae93119f : SUCCESS in 34m 56s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/5c4fba8d04a24b9a85738265d0ab0d75 : SUCCESS in 25m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e2a34e0ab66bcadf4fcaa8d0d86913d3dae398ba","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-24 10:48:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d23dd47f6b9e0b6eafcbfd96b6153e78347d28e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 12:10:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3c815d648924988a3c5f33ebae17185\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995d9414530441d6bf43055fe57f03b0 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c8f1456bd85472aa06da59f6de3b2a3 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f13d3034511b4859be428658b68d3734 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81f88db9af934a478bc2397dc2059150 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc41c5e652774528b1b5dd3d808b469a : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9610b8318ddd46df98c7a63ab9bc6136 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d70e040be8f41be9a79023de4a34565 : SUCCESS in 3m 34s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/0dd905f5eeda4e388365541c2bc1caf5 : SUCCESS in 7m 26s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/9d95dd1233054106859c7c49928b63d6 : SUCCESS in 39m 32s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/578673d9a6b44d88b76cf480181b40a6 : SUCCESS in 39m 54s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69f256b1ac6e422f9a2339ab4076b6a4 : SUCCESS in 32m 26s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/f7cca90632d84964aed38272075779a2 : SUCCESS in 45m 27s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/8e2302145a1b4b01856fcad4fad2f7ba : SUCCESS in 34m 03s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aec17ca5147a4f0e8596d4738efce48b : SUCCESS in 30m 29s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/92cd2d94f15242708202a9c2ba9ee3af : SUCCESS in 41m 02s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b96d3e608924c5692707fe58326c3fb : SUCCESS in 1h 02m 59s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/0f25b4a4dd294269a007f39f1cebbd68 : SUCCESS in 1h 20m 04s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/547a9e676a524a16ae1d90fb7e299846 : SUCCESS in 45m 26s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/1b968ff2d5ae4c9185423b97cde8e243 : SUCCESS in 36m 23s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96086adafc3d4d6da6a9b85b9993283b : FAILURE in 39m 40s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/957fff100b51496e8b4d7045c2c30153 : FAILURE in 42m 20s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b415a8d3771442cbaa5a87615a39a778 : FAILURE in 50m 14s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/29e85741e48b4462bba6e98f41812c3a : FAILURE in 49m 28s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ca418d32b304e05ace742534279c4b1 : SUCCESS in 45m 02s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/d5c5d01f67a14c468ff11a5189229a74 : SUCCESS in 42m 42s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/bc4092e1f9224b779bf6ffc9897170ba : SUCCESS in 37m 46s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/987bc14dea60448aa9983ec751658d36 : SUCCESS in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"00a1a53cbb47e3ea0a45aa9bb90c5d41d6c6efa7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-24 12:36:52.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"2de4f48c66c3ec993dba0e619e08973735b42247","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-05 15:45:09.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"38830eeaebd831bcdd5ad0c0e834de175b3a16d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 15:45:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ad1ed1a0bf134fad4bde19214feaf26a31ede4f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 17:05:37.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5769b7608df04b08898e75471a6f9884\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c50ed186ebd420db319aae1a4a52ad2 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/857ab5f1d2b94119ae373b8d67a54d26 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2313acd5dff440bda0c7446d702d60e7 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d10323241f24c43a585f6a6be814b37 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88bcccd8c7bf40428ba758f36ed69250 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc703c8813cc4531b87ce0018fabab4b : SUCCESS in 3m 25s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/9158ff365caf4bac96fc401c51f404d4 : SUCCESS in 6m 44s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/5a876c78532a45c0a9b4c16189ae2c8c : SUCCESS in 45m 54s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2def94f6f200412489d4866b4d06e537 : SUCCESS in 33m 30s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84b9f3741760491eb2c6fa91414c541d : SUCCESS in 32m 07s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/aa3b3e5f0dd04c19889430f6b27b549b : SUCCESS in 38m 24s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/b2372e599b9a419cae6448e2e8e93e57 : SUCCESS in 33m 09s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/891413aa9556427e80bccbd6d6e7375b : SUCCESS in 31m 17s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/40408fede05b433caf3001b32fd060d2 : SUCCESS in 35m 35s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7c20846c2f947c78cad36b9768fb474 : SUCCESS in 1h 00m 24s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/356237014d9343738d76a2bb4ead9e02 : SUCCESS in 1h 19m 07s","accounts_in_message":[],"_revision_number":6},{"id":"3a942602310db616ee5bc0295713b542e4fa0161","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 17:05:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"98353fb94ce143da8f95138e786f07760557e37a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 17:07:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c91565af7e2449158e76e608f93278e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbfc20f01d24a1281a8196c794e9963 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e87bb66bfd144063aeea78ee6ff8b5a4 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a6ce28314677557427aec77053c0107923822b70","revisions":{"17c595bb01687cd25710133b38240b177b2d1e18":{"kind":"REWORK","_number":1,"created":"2023-05-23 14:25:35.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/43/883943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/43/883943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/43/883943/1"}}},"commit":{"parents":[{"commit":"7979498316762b51759ef056fabd517c1f0909cd","subject":"Skip unnecessary SDK get_machine calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7979498316762b51759ef056fabd517c1f0909cd"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-05-03 10:10:31.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-05-23 14:24:47.000000000","tz":60},"subject":"Replace Cirros references with generic descriptors","message":"Replace Cirros references with generic descriptors\n\nOften it is beneficial to use the functionality provided by `use_cirros`\nand `cirros_deploy_image_upstream_url` to download and test with images\nof other OS distributions. This patch changes the variable names to\nreflect that use case.\n\n`use_cirros` -\u003e `download_custom_deploy_image`\n`cirros_deploy_image_upstream_url` -\u003e `custom_deploy_image_upstream_url`\n\nThe functionality of the old variables is retained for backwards\ncompatability.\n\nCo-Authored-By: Alex Welsh \u003calex@stackhpc.com\u003e\n\nChange-Id: I8b3a973bd5ef19333af563c170b2a28fbbfb5c76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/17c595bb01687cd25710133b38240b177b2d1e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/17c595bb01687cd25710133b38240b177b2d1e18"}]},"branch":"refs/heads/master"},"e309db9786b6b00c27e559fa293fcc988cf185b0":{"kind":"REWORK","_number":2,"created":"2023-05-30 09:14:23.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/43/883943/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/43/883943/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/43/883943/2"}}},"commit":{"parents":[{"commit":"7979498316762b51759ef056fabd517c1f0909cd","subject":"Skip unnecessary SDK get_machine calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7979498316762b51759ef056fabd517c1f0909cd"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-05-03 10:10:31.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-05-30 09:14:12.000000000","tz":60},"subject":"Replace Cirros references with generic descriptors","message":"Replace Cirros references with generic descriptors\n\nOften it is beneficial to use the functionality provided by `use_cirros`\nand `cirros_deploy_image_upstream_url` to download and test with images\nof other OS distributions. This patch changes the variable names to\nreflect that use case.\n\n`use_cirros` -\u003e `download_custom_deploy_image`\n`cirros_deploy_image_upstream_url` -\u003e `custom_deploy_image_upstream_url`\n\nThe functionality of the old variables is retained for backwards\ncompatability.\n\nCo-Authored-By: Alex Welsh \u003calex@stackhpc.com\u003e\n\nChange-Id: I8b3a973bd5ef19333af563c170b2a28fbbfb5c76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e309db9786b6b00c27e559fa293fcc988cf185b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e309db9786b6b00c27e559fa293fcc988cf185b0"}]},"branch":"refs/heads/master"},"53d8d1e25b818cdbf7c28cb05fdc6d609a3df30a":{"kind":"REWORK","_number":3,"created":"2023-05-31 13:41:40.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/43/883943/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/43/883943/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/43/883943/3"}}},"commit":{"parents":[{"commit":"6a6f0732b0ba16fe0e97e2bd41bd867fe5a87b54","subject":"Merge \"Set [inspector]require_managed_boot explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6a6f0732b0ba16fe0e97e2bd41bd867fe5a87b54"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-05-03 10:10:31.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-05-31 13:40:44.000000000","tz":60},"subject":"Replace Cirros references with generic descriptors","message":"Replace Cirros references with generic descriptors\n\nOften it is beneficial to use the functionality provided by `use_cirros`\nand `cirros_deploy_image_upstream_url` to download and test with images\nof other OS distributions. This patch changes the variable names to\nreflect that use case.\n\n`use_cirros` -\u003e `download_custom_deploy_image`\n`cirros_deploy_image_upstream_url` -\u003e `custom_deploy_image_upstream_url`\n\nThe functionality of the old variables is retained for backwards\ncompatability.\n\nCo-Authored-By: Alex Welsh \u003calex@stackhpc.com\u003e\n\nChange-Id: I8b3a973bd5ef19333af563c170b2a28fbbfb5c76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/53d8d1e25b818cdbf7c28cb05fdc6d609a3df30a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/53d8d1e25b818cdbf7c28cb05fdc6d609a3df30a"}]},"branch":"refs/heads/master"},"fbcd5d727170c9e09f8323f8d1e6445e22457f62":{"kind":"REWORK","_number":4,"created":"2023-06-21 15:02:54.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/43/883943/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/43/883943/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/43/883943/4"}}},"commit":{"parents":[{"commit":"d52ef0667ed907f64dbc523ab832db371632d41b","subject":"Merge \"Reduce multiple tasks to task with loop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/d52ef0667ed907f64dbc523ab832db371632d41b"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-05-03 10:10:31.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-06-21 15:00:41.000000000","tz":60},"subject":"Replace Cirros references with generic descriptors","message":"Replace Cirros references with generic descriptors\n\nOften it is beneficial to use the functionality provided by `use_cirros`\nand `cirros_deploy_image_upstream_url` to download and test with images\nof other OS distributions. This patch changes the variable names to\nreflect that use case.\n\n`use_cirros` -\u003e `download_custom_deploy_image`\n`cirros_deploy_image_upstream_url` -\u003e `custom_deploy_image_upstream_url`\n\nThe functionality of the old variables is retained for backwards\ncompatability.\n\nCo-Authored-By: Alex Welsh \u003calex@stackhpc.com\u003e\n\nChange-Id: I8b3a973bd5ef19333af563c170b2a28fbbfb5c76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbcd5d727170c9e09f8323f8d1e6445e22457f62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/fbcd5d727170c9e09f8323f8d1e6445e22457f62"}]},"branch":"refs/heads/master"},"cb385ce90c6b288d8acf26cac3512c6dacf98339":{"kind":"REWORK","_number":5,"created":"2023-08-09 16:12:47.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/43/883943/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/43/883943/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/43/883943/5"}}},"commit":{"parents":[{"commit":"de39802cdb75e5eeba0895824f83e6aba7420ffd","subject":"Merge \"Allow nginx to access images again on systems with SELinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/de39802cdb75e5eeba0895824f83e6aba7420ffd"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-05-03 10:10:31.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-08-09 16:12:40.000000000","tz":60},"subject":"Replace Cirros references with generic descriptors","message":"Replace Cirros references with generic descriptors\n\nOften it is beneficial to use the functionality provided by `use_cirros`\nand `cirros_deploy_image_upstream_url` to download and test with images\nof other OS distributions. This patch changes the variable names to\nreflect that use case.\n\n`use_cirros` -\u003e `download_custom_deploy_image`\n`cirros_deploy_image_upstream_url` -\u003e `custom_deploy_image_upstream_url`\n\nThe functionality of the old variables is retained for backwards\ncompatability.\n\nCo-Authored-By: Alex Welsh \u003calex@stackhpc.com\u003e\n\nChange-Id: I8b3a973bd5ef19333af563c170b2a28fbbfb5c76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cb385ce90c6b288d8acf26cac3512c6dacf98339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cb385ce90c6b288d8acf26cac3512c6dacf98339"}]},"branch":"refs/heads/master"},"a6ce28314677557427aec77053c0107923822b70":{"kind":"REWORK","_number":6,"created":"2023-08-10 10:41:39.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/43/883943/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/43/883943/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/43/883943/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/43/883943/6"}}},"commit":{"parents":[{"commit":"de39802cdb75e5eeba0895824f83e6aba7420ffd","subject":"Merge \"Allow nginx to access images again on systems with SELinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/de39802cdb75e5eeba0895824f83e6aba7420ffd"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-05-03 10:10:31.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-08-10 10:41:32.000000000","tz":60},"subject":"Replace Cirros references with generic descriptors","message":"Replace Cirros references with generic descriptors\n\nOften it is beneficial to use the functionality provided by `use_cirros`\nand `cirros_deploy_image_upstream_url` to download and test with images\nof other OS distributions. This patch changes the variable names to\nreflect that use case.\n\n`use_cirros` -\u003e `download_custom_deploy_image`\n`cirros_deploy_image_upstream_url` -\u003e `custom_deploy_image_upstream_url`\n\nThe functionality of the old variables is retained for backwards\ncompatability.\n\nCo-Authored-By: Alex Welsh \u003calex@stackhpc.com\u003e\n\nChange-Id: I8b3a973bd5ef19333af563c170b2a28fbbfb5c76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a6ce28314677557427aec77053c0107923822b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a6ce28314677557427aec77053c0107923822b70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
