)]}'
{"id":"openstack%2Fbifrost~897204","triplet_id":"openstack%2Fbifrost~master~I239e06d6876de0ffbe3ca2d5fef1349045012b12","project":"openstack/bifrost","branch":"master","topic":"add_logrotate","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-10-03 14:19:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I239e06d6876de0ffbe3ca2d5fef1349045012b12","subject":"Fix Keystone default log location typo","status":"MERGED","created":"2023-10-03 08:37:03.000000000","updated":"2023-10-03 14:20:45.000000000","submitted":"2023-10-03 14:19:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897204-add_logrotate","meta_rev_id":"f9ab14dd4b0ef3bf68dcaed88957573a204c9133","_number":897204,"virtual_id_number":897204,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-03 14:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-10-03 12:15:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-10-03 08:48:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2023-10-03 08:52:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-03 12:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-03 12:22:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-03 08:48:17.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-10-03 08:52:10.000000000","updated_by":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"reviewer":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"state":"REVIEWER"},{"updated":"2023-10-03 10:14:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-03 12:15:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-10-03 12:22:02.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"0108e2c5d96c691bf928b8b54a03eae6aed08bfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-03 08:37:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"361f898120d9c898ff882de3282e65588c53698b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-03 08:37:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e0e13785a1a80959e2479f374f7ed26da2d2e97a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-03 08:38:51.000000000","message":"Topic add_logrotate-stable removed","accounts_in_message":[],"_revision_number":2},{"id":"3c75419ba4cf678f9d3f95962eb424c7f843a889","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-03 08:38:57.000000000","message":"Topic set to add_logrotate","accounts_in_message":[],"_revision_number":2},{"id":"ae95bae9bb7e8f1b290dbe408bb3fcf16fcc01a0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-10-03 08:48:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e22c4d360d8c61a1d6678dbda4e05e7f96ec1195","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-10-03 08:52:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6fcd17046a15da359c9d8c243503c5ff8c6eb47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 10:14:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbcff67c430349d99a02ac4b936cc6c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22da32e744fd481e9f37022b504474a2 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35ab9097f0bc409fb3aac017e245ce0e : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa8502537a254a8d89c5161f8a96ee33 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f970233e44bc4d748e3c7587d930dbe2 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e96d8b0ed14f47b58b0119327b74bf3a : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5510e4937c3f4240a98b20733970ca98 : SUCCESS in 5m 49s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/4a88824bdfd54eeab3e3913bd3e62c67 : SUCCESS in 8m 06s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/dd5f5f058ad04e1c8ae45e1db6ec2904 : SUCCESS in 35m 32s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28153fa009dd4780a3205a427da5a6b7 : SUCCESS in 32m 58s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc2af2ee34d4422bae28e6cfa2d012db : SUCCESS in 36m 16s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/f2bb76615dd44210abae649ff655170c : SUCCESS in 36m 26s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/25f849b1743f49b8a1b96f50e38f16a5 : SUCCESS in 31m 37s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b057e5b96c9b46db9d7be33facbf488b : SUCCESS in 31m 54s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/367eccfe1c944c6daa80074126d0b084 : SUCCESS in 32m 10s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90a15b2507a34f47870ed5ce2c81744b : SUCCESS in 1h 12m 15s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/0058b5abb957434f8075f002e2d099b9 : SUCCESS in 1h 31m 13s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cfeba9c701a4ccfbb6f19ef7f329899 : SUCCESS in 43m 54s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/67b17de1d142429a9d2ba2798c38a4e2 : SUCCESS in 44m 44s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be98fdd9248e43859c9462f1f9e0cdfc : FAILURE in 40m 42s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/25f30055f710492eafee55bb2953470a : FAILURE in 41m 58s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01b64066bda24347997bb214b95bc258 : FAILURE in 47m 47s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/800e8f4f4742482abc39dbc6d17245c2 : FAILURE in 55m 26s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de467e2405094995bdc5ff4ca7fe90e8 : SUCCESS in 42m 11s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/60ee3fb7f8384b6a8c8c9fa9347848e2 : SUCCESS in 46m 48s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/1d940e28e3164534a4e15b15afd8d47e : SUCCESS in 39m 26s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/2030353eedcf459681c129cda1c4d685 : SUCCESS in 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed9d6d7ec4f3bd73077b63203835b8898865f176","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-03 12:15:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1faeff8e7512000fad517e14a6e317e0205a9e26","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-10-03 12:22:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0c6b67d5d0fd19d1d15e4fe4e1d33e342bd16973","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:22:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e5cf8d59a7b94ed6b23f3e5769a5c041c1d73f05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 14:19:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2ee79b7c31f4da8a73e1f232da1e5f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57213b49985f4eb6a343670b9f492717 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88bad754c4a54c379d3742bbf2d88888 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/066fc6b60d3143e48b650add8950ebd9 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8204a9c594ff4d899f4e03169767c0df : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/209f2cec6e284be892ab93bf763f686d : SUCCESS in 6m 13s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/859a1f645b164fec8dd957ecc424cb78 : SUCCESS in 7m 22s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/7488ac30b9384c70ba1113a6f399a789 : SUCCESS in 45m 16s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/703cce5e3df141b6b3ba849290c1aacc : SUCCESS in 44m 28s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97ace858f0c3408583f9667ed254d08f : SUCCESS in 32m 39s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/b7c00b4c4cf849f7a18a85c6c03398ec : SUCCESS in 47m 43s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/21bf675529524b5a921e32ed6b339125 : SUCCESS in 41m 46s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9dab623a03d4a9b9cc3c57901eb6d79 : SUCCESS in 29m 48s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/112a5ccee3da42ad9430849da2c4284b : SUCCESS in 47m 16s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f7e29176d854ca7bcfff68c9df3dfce : SUCCESS in 1h 10m 52s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/6d08055ea9fb4db9b4b131737cb320ba : SUCCESS in 1h 52m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1d3ad755509da85281ddbf80339fcd5712944ce6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 14:19:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f9ab14dd4b0ef3bf68dcaed88957573a204c9133","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 14:20:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797f3fda0e2f40b98d7e0aa6f5908757\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/176ad2e17cbe4e559ca8d64165283bc2 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca44e8480500963496a561b1b34aec5aacc68aa2","revisions":{"e5241f85997e843c6b4cc5cb8fab189cdef96546":{"kind":"REWORK","_number":1,"created":"2023-10-03 08:37:03.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/04/897204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/04/897204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/04/897204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/04/897204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/04/897204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/04/897204/1"}}},"commit":{"parents":[{"commit":"912893ae83e6d60e8067dc1509d77b4c8cc80a32","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/912893ae83e6d60e8067dc1509d77b4c8cc80a32"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-10-03 08:34:36.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-10-03 08:36:05.000000000","tz":180},"subject":"Fix Keystone default log location typo","message":"Fix Keystone default log location typo\n\nFollowup on I7c2ba2b263dd84641674b60aed8971b129aa846f fixes Keystone\ndefault log location typo.\n\nTrivialFix\n\nChange-Id: I239e06d6876de0ffbe3ca2d5fef1349045012b12\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e5241f85997e843c6b4cc5cb8fab189cdef96546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e5241f85997e843c6b4cc5cb8fab189cdef96546"}]},"branch":"refs/heads/master"},"ca44e8480500963496a561b1b34aec5aacc68aa2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-03 08:37:53.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/04/897204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/04/897204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/04/897204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/04/897204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/04/897204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/04/897204/2"}}},"commit":{"parents":[{"commit":"912893ae83e6d60e8067dc1509d77b4c8cc80a32","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/912893ae83e6d60e8067dc1509d77b4c8cc80a32"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-10-03 08:34:36.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-10-03 08:37:44.000000000","tz":180},"subject":"Fix Keystone default log location typo","message":"Fix Keystone default log location typo\n\nFollowup on I7c2ba2b263dd84641674b60aed8971b129aa846f fixes Keystone\ndefault log location typo.\n\nTrivialFix\n\nRelated-Bug: #2033528\nChange-Id: I239e06d6876de0ffbe3ca2d5fef1349045012b12\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/ca44e8480500963496a561b1b34aec5aacc68aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/ca44e8480500963496a561b1b34aec5aacc68aa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
