)]}'
{"id":"openstack%2Fbifrost~903755","triplet_id":"openstack%2Fbifrost~master~I9b171809ea7f013bb37a3f7b9669150f8f8a0a26","project":"openstack/bifrost","branch":"master","topic":"collect-dnsmasq-config","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-08-06 14:34:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9b171809ea7f013bb37a3f7b9669150f8f8a0a26","subject":"[DNM] collect dnsmasq config","status":"ABANDONED","created":"2023-12-15 14:08:55.000000000","updated":"2024-08-06 14:34:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"16294c9af1aa177134c2782b97831661b6cb4ecd","_number":903755,"virtual_id_number":903755,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-15 17:39:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-15 15:53:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ab9ab56d8d3d86c1812a0b209a737fde385e512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-12-15 14:08:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad6e8af9afdaab9537b83816b64bb4efa1e887c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 15:53:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92361a6c442a4a4badda165b3cf0a1ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788e2c1b700a471fa539237f00d25ac2 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38175d514a5341638b2fe1a513cd4b83 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f94390fcd6434c8da2ebdebe6440af27 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad978fc9e68d49bd86257319a17946b7 : SUCCESS in 5m 38s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/63869f27aab0475ca4972025270fd244 : SUCCESS in 6m 39s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/77910a79a5114d6181bc1d6f0b93825c : SUCCESS in 45m 42s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/066e4eb89aca4a7a94c8ba5b80a93514 : SUCCESS in 37m 20s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db2e53cebd7b4887bb55ee54c6ad6af7 : SUCCESS in 39m 53s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/f38aa2e6c1cc4b8e9c6229b51a8e52ff : SUCCESS in 36m 46s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/5499e674d0eb4ec280a7a350666acf3b : SUCCESS in 34m 59s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/52c4be99d1e243fc82d7221680fa2044 : SUCCESS in 34m 02s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8f6c62c0fbc42f8b27a1bea66098bf0 : SUCCESS in 36m 43s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/e4dbacdbb88d45579301389e02a31060 : SUCCESS in 30m 09s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/8ab1b52559c54c62a2ad1e8bd38f57ca : SUCCESS in 31m 10s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50f762f27957428fae3a2c7f01032405 : SUCCESS in 1h 07m 56s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/598ae69387ff43958ae5ab1a6e591547 : SUCCESS in 1h 37m 46s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bac125eddce24027bef87284da060c34 : SUCCESS in 44m 43s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/cb93be1b0469443f86cdbaf43108fa2d : SUCCESS in 42m 42s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c049dfa64924e8ab0ee5364d8d2f063 : FAILURE in 9m 11s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/cc5009b6884843deba5f05c2f56959aa : FAILURE in 43m 11s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d259202da2464d31b633b084faf6d65b : SUCCESS in 42m 51s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/9ad8ea2a050d436c8f8a56321149b29b : FAILURE in 22m 56s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb741686be4c4dd8af78cbf0c38ec25b : SUCCESS in 41m 12s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/73647d2a12bb4e73bacbff1a1f46618e : SUCCESS in 44m 35s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/1260bb30f60545e69cb4cf6c8edd62bc : SUCCESS in 36m 14s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/673ebd7de2894b12816e9085207385d7 : SUCCESS in 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7d3ae315199ca500802715787c37acdf42bd59f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-12-15 16:04:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6e1e30424d6e15cf75026450feacefe1840e57a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 17:39:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d67d278eba5146e79036dcaab24e1876\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f22cdcea18d6412cb2c2e384c1a89ee5 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4acad9d2bc241ffa3711723563ab164 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b779c8239ace4881b4e800bada72a677 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1bdf9e47624c4a9d9ec85fd08fd5d9 : SUCCESS in 5m 50s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/6320557a1e6b4edb8d0864d03bebaf98 : SUCCESS in 8m 40s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/ddcbc3a5a3234417b509c3eaf834de55 : SUCCESS in 45m 21s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2a1b80f6e66477bbb6bfe1cc164ef88 : SUCCESS in 43m 13s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbfec494e8c94433a458e0774ad5c66b : SUCCESS in 46m 35s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/98dacbc1c3664ee8b4ffddb6505993ae : TIMED_OUT in 1h 00m 48s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/760320f82f4c4fa387105a9d78cf44cc : SUCCESS in 30m 26s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/2e457d4463994ff6a130245c07685886 : SUCCESS in 34m 59s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ff7776833604e5da96f321ad13f1082 : SUCCESS in 42m 29s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/bf07179d8b2247b78f2c2cf6f6d0c709 : SUCCESS in 47m 31s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/5b2f62fb24be4e6dab8304c5497e7aeb : SUCCESS in 38m 20s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b51cbc8ddcb546eca4ffeedf85a30469 : SUCCESS in 57m 54s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/088b8d6e9df848db97ccd9e443e221b9 : SUCCESS in 1h 27m 31s\n- bifrost-integration-tinyipa-keystone-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efe0de3e34424448bb4efa0b6d158186 : SUCCESS in 40m 33s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/32baf84999084062a89b67732a1ef662 : SUCCESS in 36m 56s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d9ce7d41891485ead4745c6c713e294 : FAILURE in 8m 58s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/6f12f854e0e642e6bd2926c25fd73035 : FAILURE in 44m 10s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de56225f54d540dba08e5ab48b41bf19 : SUCCESS in 45m 33s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/3bc1373c7c7c42b8b9fe4ead5acb2dfe : FAILURE in 19m 19s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e7491aac0874c1cb2d71a78b9a048dd : SUCCESS in 41m 28s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/f448f4efcd1a4e609284f0526708d1b5 : SUCCESS in 44m 25s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/d2bc9b226f634671b1fba0cf5c0f595f : SUCCESS in 49m 29s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/f0a32e4556de4af490b4bc1e641420c3 : SUCCESS in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16294c9af1aa177134c2782b97831661b6cb4ecd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-08-06 14:34:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b742239a274c83963d762e31fc6ac6bf9dcd3e10","revisions":{"4ab517b8f17e6dfcadca66cd562820958098e6e1":{"kind":"REWORK","_number":1,"created":"2023-12-15 14:08:55.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/55/903755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/55/903755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/55/903755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/55/903755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/55/903755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/55/903755/1"}}},"commit":{"parents":[{"commit":"83066af17d8a90d96df87a4bd8dc1437465f0eb7","subject":"Merge \"Use sha256 instead of md5 by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/83066af17d8a90d96df87a4bd8dc1437465f0eb7"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-12-15 14:08:42.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-12-15 14:08:42.000000000","tz":60},"subject":"[DNM] collect dnsmasq config","message":"[DNM] collect dnsmasq config\n\nChange-Id: I9b171809ea7f013bb37a3f7b9669150f8f8a0a26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4ab517b8f17e6dfcadca66cd562820958098e6e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4ab517b8f17e6dfcadca66cd562820958098e6e1"}]},"branch":"refs/heads/master"},"b742239a274c83963d762e31fc6ac6bf9dcd3e10":{"kind":"REWORK","_number":2,"created":"2023-12-15 16:04:30.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/55/903755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/55/903755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/55/903755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/55/903755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/55/903755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/55/903755/2"}}},"commit":{"parents":[{"commit":"83066af17d8a90d96df87a4bd8dc1437465f0eb7","subject":"Merge \"Use sha256 instead of md5 by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/83066af17d8a90d96df87a4bd8dc1437465f0eb7"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-12-15 14:08:42.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-12-15 16:04:24.000000000","tz":60},"subject":"[DNM] collect dnsmasq config","message":"[DNM] collect dnsmasq config\n\nChange-Id: I9b171809ea7f013bb37a3f7b9669150f8f8a0a26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b742239a274c83963d762e31fc6ac6bf9dcd3e10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b742239a274c83963d762e31fc6ac6bf9dcd3e10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
