)]}'
{"id":"openstack%2Fbifrost~917537","triplet_id":"openstack%2Fbifrost~master~I1093e87048b5cab51add180b90be836628a15731","project":"openstack/bifrost","branch":"master","topic":"no-focal","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2024-05-13 16:03:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1093e87048b5cab51add180b90be836628a15731","subject":"Remove last traces of ubuntu focal","status":"MERGED","created":"2024-04-30 06:59:04.000000000","updated":"2024-05-13 16:06:06.000000000","submitted":"2024-05-13 16:03:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917537-no-focal","meta_rev_id":"295851ea9dc733548543cc28cca48b2337e4c986","_number":917537,"virtual_id_number":917537,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-13 16:03:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2024-05-13 14:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2024-04-30 07:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2024-05-13 14:24:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 07:32:52.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-04-30 08:28:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 14:24:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6b3f1cd48d3f7ee55b37d27ec0c7c9e14f203417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-04-30 06:59:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df419ff87b64798acd5562e90993032a1fa43a5c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-04-30 07:32:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"acc911804ff3c3fa92e4988b063e3f0f214eec24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 08:28:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2ae42dac7424ef0a053bb725ee6b8db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8564c374f69b45429c185e6887afd4c3 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39706277f64143a08e019c0a6ef3dba5 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a3f5ba9e37a4686ad6ad94a2730b847 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25aa06e602ba4daeb3474649a3e25a90 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5984746742844141a3915e9024a395cb : SUCCESS in 4m 51s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/67089d0aa3924c2a973336a0b332e9e0 : SUCCESS in 6m 29s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/51d5d8967b504450912aa50ec3c7e9b9 : SUCCESS in 43m 00s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/530a5ee059844f93813ca6966b25b200 : SUCCESS in 33m 45s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d52ed8520bb545f5aea510ef8b8b325a : SUCCESS in 43m 12s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/274fd8d73e17458b81cd8983053f31da : SUCCESS in 49m 20s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/dbd122b59a15487591d8470e80416265 : SUCCESS in 36m 07s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/5d3c8f978e1a4dba8ae94345a6a34532 : SUCCESS in 35m 16s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15f7d592f8394de0bd652314b671ca94 : SUCCESS in 45m 22s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/971eeb3a2c80499497be3111324dd1dd : SUCCESS in 35m 47s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/37d675e6028a411d805732c58b67cfdf : SUCCESS in 46m 08s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/5d135c09b2ad4815890a35e39bf12bf5 : SUCCESS in 43m 12s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60ddb16bc24c44fe88bfe19a58a134f4 : SUCCESS in 41m 03s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afeb80ea731543d7a19345f6b3dfa83e : SUCCESS in 1h 22m 56s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/4b2d9c40545a42d5b768b719e8c32b49 : SUCCESS in 1h 26m 30s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/404797f894eb42c6ad37b20f26040980 : SUCCESS in 1h 19m 08s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/5eee2120097d4bdb811525bf11742228 : SUCCESS in 33m 47s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05d2ae569b6146dfa0a50515e7381b28 : SUCCESS in 40m 26s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/5399d2245b874fb7abeee2c8afa7471c : SUCCESS in 38m 07s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bc6b5036f7848f789ede350bc1ede4a : FAILURE in 40m 30s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/300f190832204d8c870456141a5c547a : SUCCESS in 42m 56s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1675c11c154c410594ff9910d52019e7 : SUCCESS in 38m 48s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/37a7c4e1069c4701a32f7d6592dff89c : SUCCESS in 34m 18s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/b48397e1bcef4b4790a5f170feb4346f : SUCCESS in 33m 08s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/b394e6d4fa1a409db4539f4ed1aaf7a1 : SUCCESS in 22m 04s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/45e8a210e67544e7a474920d7951b25a : FAILURE in 2m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90901b4d06007c8a8cfca84b578e1b78490616b4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-05-13 14:24:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d4c486c86155f6f4ecab9418f028ecc4605e9b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 14:24:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"332c0280e6c3a84a029faef92dc98ceae725d096","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:03:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/866226e3af084f0fa359f0b1cd3c22cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3bf7c1cabd8406a8e7bc0e9a8def657 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc3f91c413264a81997c7014109bd672 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c136fb14778c4ac29f9402ec394dd021 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb1499f662042a69b764e40f1943b52 : SUCCESS in 5m 19s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/e7d96db7945140bb83bc944735162c7a : SUCCESS in 7m 19s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/35bb3f2f06214401a180e9401423c84f : SUCCESS in 46m 47s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a82f4f447e684bbeb6f545353a492c54 : SUCCESS in 45m 38s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ecd59ac858e4540b5bddd4781eb82f4 : SUCCESS in 46m 39s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/487cc93c90994f048b2b0c3330b20dc0 : SUCCESS in 44m 20s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/cfe52d10984047fabfe641a96f4a6f1e : SUCCESS in 47m 55s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/fdf4540d3b7746ac82f0958c6510a0c8 : SUCCESS in 49m 52s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeca4b5c7bbd4cd885401a399e664acb : SUCCESS in 39m 13s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/4f6de9eacf7d4477976e7ae3e991c075 : SUCCESS in 36m 55s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/178a7e3f643644ecacd335df23b543fa : SUCCESS in 44m 38s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/49241b3c8eb44c8f9fe9479cd309dab2 : SUCCESS in 40m 25s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dedb6819cd7b4f03b79e6406196e9bb4 : SUCCESS in 37m 49s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a69d0183ad7470499fb2932c4e24174 : SUCCESS in 1h 29m 28s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/f1fa190016794dfca8a7d67f30c22b45 : SUCCESS in 1h 29m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9b1ecc2f5ccc9938b44d3303dc8ff81b3168e4e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:03:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"295851ea9dc733548543cc28cca48b2337e4c986","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:06:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec0357aedaa94e28a965710cb9b3a431\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc876eff7b54772b66631b294b0457e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24686d322895c48c68bab9aba80b0f40bdf8d973","revisions":{"24686d322895c48c68bab9aba80b0f40bdf8d973":{"kind":"REWORK","_number":1,"created":"2024-04-30 06:59:04.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/37/917537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/37/917537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/37/917537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/37/917537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/37/917537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/37/917537/1"}}},"commit":{"parents":[{"commit":"b47833ca9e781fd3cc80d6bac848acaca1c86c8b","subject":"Merge \"CI: Add a non-voting SLURP upgrade job for Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b47833ca9e781fd3cc80d6bac848acaca1c86c8b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-04-30 06:58:21.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-04-30 06:58:21.000000000","tz":120},"subject":"Remove last traces of ubuntu focal","message":"Remove last traces of ubuntu focal\n\nThe CI jobs are no longer used by anything, and bifrost does not support\nfocal either already.\n\nChange-Id: I1093e87048b5cab51add180b90be836628a15731\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/24686d322895c48c68bab9aba80b0f40bdf8d973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/24686d322895c48c68bab9aba80b0f40bdf8d973"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
