)]}'
{"id":"openstack%2Fbifrost~917927","triplet_id":"openstack%2Fbifrost~master~Ic23dad7a0371c1cc68aa5f870225d6301ed4ed8e","project":"openstack/bifrost","branch":"master","topic":"support-setting-boot_interface-in-groupvars","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2024-05-03 20:48:36.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-05-03 07:38:24.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"hashtags":[],"change_id":"Ic23dad7a0371c1cc68aa5f870225d6301ed4ed8e","subject":"Make boot_interface customisable","status":"MERGED","created":"2024-05-02 08:07:57.000000000","updated":"2024-05-03 20:49:36.000000000","submitted":"2024-05-03 20:48:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"917927-support-setting-boot_interface-in-groupvars","meta_rev_id":"4f4cf030f7dd9e2daba38b2740fcfae8daf01be5","_number":917927,"virtual_id_number":917927,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-03 20:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2024-05-03 07:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-03 19:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-03 19:24:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-02 08:31:34.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2024-05-02 10:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-03 07:38:24.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-05-03 19:24:35.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"d5f5560fd4e770b52d5d690524703d486f871d81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2024-05-02 08:07:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6087887864692f2d20b5997752cb17ca4b27820b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2024-05-02 08:26:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d59cba51ff1b1845cc6fd38fe6307a1030f50a6c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-05-02 08:31:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dacc718daa77003a023836877108c03f1a46491d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 10:04:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/757a827e7aeb43e0847d16379b1228a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca951fa5305f488fb277b6fd731e4766 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8590164447f34480a2aa7d2e60b36831 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7c2ab09cbd0450c87c17d94ed0e9613 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad9eebad95f64c548e9dd6b10646eb5d : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d017b9fe20bf41e9ab2ae5f1635d0da1 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24d223c04134489bae44d1b9278d974d : SUCCESS in 4m 24s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/36c7eb39371449b2be43a3f22cde830a : SUCCESS in 8m 55s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/bb082bb649db4296ada21ed9180b7f6f : SUCCESS in 39m 42s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1a885ef967d4dd68f0b34c921c37fd2 : SUCCESS in 45m 13s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c4d4dc95402471eb66a69a6f12727c9 : SUCCESS in 40m 41s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/67c23739efa648e5a4d34646b5106af4 : FAILURE in 56m 56s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/737d831726a64da6bffbd5f6fb3483cf : SUCCESS in 48m 26s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/d881621b6d6547adabf76d0b313c7162 : SUCCESS in 49m 57s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/517c1474f2a7428d8b5afafcf98429db : SUCCESS in 45m 08s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/42dc3b094e624033807b6f7cb46e750a : SUCCESS in 43m 30s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/5613a635d40b4641ad5a6bcb1f478ec2 : SUCCESS in 45m 43s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/c314e0f125dc4393ae8cd7dfd896aed4 : SUCCESS in 44m 31s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9f268b45a13451db464161a3897184c : SUCCESS in 42m 02s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/487ea782470b464382d9a03187abf7b3 : SUCCESS in 1h 32m 15s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/23915870b182478e9bbcce96fdc1df55 : SUCCESS in 1h 19m 35s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c6fac46b8ce4a539c4fec45c29b73d5 : SUCCESS in 56m 00s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/cb8f801180e94f038722bbad035d332e : SUCCESS in 31m 59s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/673acea5b76a4aa5a619592de4551f03 : SUCCESS in 39m 41s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/7022488eba124580893639e16ded1964 : SUCCESS in 35m 28s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20250f99dbbb42e1b01af4456fb6dc95 : FAILURE in 51m 15s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/9de8a7f34760409bb7219544bc4f9793 : SUCCESS in 36m 23s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3dbde8b6aa85483288b5d68289355a76 : SUCCESS in 33m 58s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/81ce873415934868a6b4f8234fcd421a : SUCCESS in 31m 29s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/f167d0d437a14adbb42c992057c14087 : SUCCESS in 42m 32s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/319094bf4f4f49a58e92596c5bfdd05a : SUCCESS in 24m 55s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/1b2275cf16a34b3695abf55715d25e78 : FAILURE in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4cece6d043289c433078685e4db1bccbc8967691","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2024-05-02 10:27:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7615aedca61c9bc032763c2f9e89e1898aeeed5e","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2024-05-03 00:21:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e017e3b2a894f1af641b4d742f1731cbfdd44aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 02:16:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/120d9872e3fd448da1973d91b7f914f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5ad8b9f7254fc8a05ce6c2524f012c : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af6aeb72c9904fcdb5352d58d0e986f1 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c855fc5ec95e4f0eb24e8b2c2edb4617 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b5dda8a96de48198c9660d58d86aba3 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756207fe9ec9449bbf0d4b9bac588aa8 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9b483a093da4d27a15c2be3ac41680f : SUCCESS in 2m 25s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/6101c25d4be349c4a82ffaca96d99c07 : SUCCESS in 7m 45s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/6fec374d3900498fae1db63af45592b0 : SUCCESS in 46m 38s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ef26fcfadcb4e4a80481bcbdc250f51 : SUCCESS in 32m 32s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da5136c8db464de19faf9fee86ccec06 : SUCCESS in 47m 01s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/6c1f1e9a0f264d539e3963fe86a1989a : TIMED_OUT in 1h 00m 51s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/d7480f1ebba843c79e7832cf67ab63dd : SUCCESS in 30m 35s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/82b37130edef4eff8e380e6e72679bbe : SUCCESS in 48m 13s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/695b5eda4e5c46fcae02b06ae3136f1c : SUCCESS in 29m 48s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/63b9db64f1fb4e98a5e2e1b4ae7b7ddb : SUCCESS in 44m 32s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/c749ce0e3cf2436881e193119bac8d4c : SUCCESS in 30m 48s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/b35e2f3601d942a2bf3850c40d6ff83a : SUCCESS in 31m 13s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9d3d9ac5bad4b9eb211ec33754fca58 : SUCCESS in 30m 22s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e62f95fa50fd46baaa50a750eb1dd52d : SUCCESS in 1h 00m 46s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/d45fe02d064044c6a8ba2ecf07cc79be : SUCCESS in 1h 18m 12s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76111ae613cc40d9a841d1ad2cfe457b : SUCCESS in 54m 10s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/9e28a3e7e3cb4772a1fc9613937cbad3 : SUCCESS in 28m 47s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fc0316aa21b4ae6a6f2523c2440ed59 : SUCCESS in 35m 00s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/d3ae06721fef42ae980d9ea384f0d245 : SUCCESS in 43m 52s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/816a57ebf37e41cdbded846c3f45d962 : FAILURE in 45m 26s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/d0afb4cefcaa4bb19678ac506d78daaf : FAILURE in 57m 21s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3436b813be54f6ca01cb69d70444623 : SUCCESS in 47m 18s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/aab15c5e59ce4665b449d51eabe7a2e0 : SUCCESS in 40m 09s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/93435d0a00cd47f883822da14d4e0ddc : SUCCESS in 32m 49s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/62bb9f0404814785a782ba0815500569 : SUCCESS in 27m 14s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/88686de337ff4c8789ad74af9b08c41f : FAILURE in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56810739994b3d429a437429e259c8fbb54b9b9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2024-05-03 06:21:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e2ef40b2b44f44f67c91a820d2344ad8638d7e52","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2024-05-03 07:25:06.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"02c36ff379dbfdf3cae4c8c9b37265d720911bd2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-05-03 07:38:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c16844a85ff65e040636546ed161cd7d3b20a013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 08:02:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e99651a35044e90a7c87e1960c925d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d440ee553926436782dc501c596c8bae : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caaa109c84204de7a96be3c01a75264a : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb990ec9c2f6435ea8920637414c3633 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/481c9d4af8c446a3a2ca45b130b62e54 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a454e61510043cfb2f11843f5dbc62c : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6afa50ac5c7e43cb87deef5b40f4c4cf : SUCCESS in 2m 54s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/64f0428faa7147dab9f0a6736563a550 : SUCCESS in 6m 05s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/cbd17f409aab4a148d9549825c33616c : SUCCESS in 43m 35s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/406ba13787224a4fa844dd6f1a5ad47b : SUCCESS in 32m 13s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/238aa17543c74f4b9c73108c3466e173 : SUCCESS in 48m 38s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/76ce4b8c89d04da387a25d8d5f1dc6fc : TIMED_OUT in 1h 00m 50s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/56b538657a0247918c553c8b5ea0aeea : SUCCESS in 32m 09s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/fbc499584b1543699781a77316686476 : SUCCESS in 34m 39s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83244b0e3c114f65934d4e28d932b3ff : SUCCESS in 44m 06s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/773682eaa3ce4de6a2214e8a4acbbe19 : SUCCESS in 32m 49s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/3db505c3bf504e7abe44456782d1659d : SUCCESS in 47m 28s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/b80b9fe616844fbd83d7197e772e3fd7 : SUCCESS in 39m 34s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/675baca30ae34bf6b1054fb4fb68d887 : SUCCESS in 37m 39s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3b8274ed7224f55a22f47111dade2ed : SUCCESS in 1h 26m 41s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/15107b5e9cec4f2f8b82a7b48f80b328 : SUCCESS in 1h 35m 32s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9232ea1ce388477e8a11ecff244389ea : SUCCESS in 1h 21m 46s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/4c23d5a11aff4368b833732e3758b809 : SUCCESS in 31m 38s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07b672b77ef94f7a8cf8d0ea5c99dc43 : SUCCESS in 34m 05s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/909f91d949b94c8cb9b9a2099504586f : SUCCESS in 33m 48s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ca7d3a5897941be99a389726c5a479f : SUCCESS in 45m 22s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/7bf39692cb924576a1c23343100e50b9 : SUCCESS in 44m 05s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dbe27075a324e2dbae4090de4178dab : SUCCESS in 32m 36s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/8b567a11bfa641f9bbc40093d653cafe : SUCCESS in 44m 00s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/9f918ec860cc4868a298bdcf13430e7c : SUCCESS in 48m 26s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/66118d570b76491989da5d6537437675 : SUCCESS in 23m 21s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/e7ff28b5e9344397b7566b35d0203110 : FAILURE in 2m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4fb7d452a8ff3169724887ca50a561d29d5a70b1","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2024-05-03 08:29:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9aeb066ca51ea777cdc90bf5c8fe1dd74b8e609c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 09:55:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f2b1013c95244d08cd6c75efc49f63e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/990e5a26f22042c1a160769db6e8ac7b : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e134561a0ddc47378b6dbe9ba9d4054e : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b89e992682d42d2ba13fe8b75f5b1bd : SUCCESS in 5m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c150deec9cf64a45b904f80d835fd8a5 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28383d6a7db04c3b9fb77719c19016ec : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d559c98f32c400aa21f9b46d593ce61 : SUCCESS in 3m 43s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/47c6cd6b853649fd8e520b7458ea6fa4 : SUCCESS in 6m 37s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/0fb961e7552a4d01ae06e1fcd06e082e : SUCCESS in 37m 02s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd42eccbf819471883c1ee6729e264e2 : SUCCESS in 36m 39s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c44b5f8dc3049c8b7b8226b622d0d6c : SUCCESS in 33m 29s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/1e13dd00cc8b4830bd51a13906f16a97 : SUCCESS in 36m 36s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/a4d2575595f64b91ac8f57ebc3814314 : SUCCESS in 37m 14s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/a83c5821f0224f4da2a0227b4658757a : SUCCESS in 39m 40s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f66ab37893c64537b9f80f3118697f35 : SUCCESS in 30m 34s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/0b02b023a48d4b55878798d970adeabc : SUCCESS in 32m 36s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/8e8dee3929de4a8f8c0d0a0584d4c2c9 : SUCCESS in 44m 42s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/3e5b2a981a6c48a780471c1ef8b50e8e : SUCCESS in 37m 06s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/094549a4f78d40399ced8c2bb48eb9ef : SUCCESS in 40m 25s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e9d603b074d421d97c271715ffed221 : SUCCESS in 1h 24m 40s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/146d2226e3ef4a9286e0361b26483de0 : SUCCESS in 1h 10m 50s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87c15fe19e1c4676810fe629ffa98f43 : SUCCESS in 1h 05m 35s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/4d0763aa6817487ea58551f941c24c3c : SUCCESS in 31m 33s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8964600469e4eaaa07fec7992334f6a : SUCCESS in 43m 23s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/ec4dde6c335a40e89e6f040f66ef0c6f : SUCCESS in 45m 02s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/107e41853eb145469c27f632fcc2961b : SUCCESS in 34m 02s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/bf9bbbbc3ea947458a8e5a9e583fdd4c : SUCCESS in 41m 22s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37f5ff2421174a3bb156657bd767e7af : SUCCESS in 44m 26s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/a3e17e3647064eca84f1724283c96506 : SUCCESS in 48m 22s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/e1b9e911fe8d4553bfd84867ba2b4391 : SUCCESS in 32m 48s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/59bb91e7a9d843008c0cbf008f76fcd3 : SUCCESS in 27m 38s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/e948cd0ab6a94ca085c444a26b151eaa : FAILURE in 2m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1b77a2cf127794bc0bac5d2e8ad720f9f92308b2","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2024-05-03 19:24:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c915e06cfcbd1640a93f42c947704fc6d6b6e9c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 19:25:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b1249c566353bbd606b07476ad53875dcda81ec0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 20:48:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58236a163acf421bb5cbc010b0deb5ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0ee55112234b9bb2ba2c1457c6d370 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6181a1a1a236441c8d6426b3503fad82 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9e1b5ce1acf4978b5ab1b5c89dd787f : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f3c2b6fcc048beb399b8b12ec380da : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/734f3d8264254c65bcf8e7bb46f8d088 : SUCCESS in 2m 39s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/3b1a6cf972cc4e2fb06a5ab0501ffa01 : SUCCESS in 7m 26s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/ee396ccafad74f79adbaf5a8b31dacc2 : SUCCESS in 45m 41s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a31c1d6aa6044a38cfd079fde948e98 : SUCCESS in 39m 58s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5563f8e2edc941b7b57df0230d754b39 : SUCCESS in 42m 12s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/2f3a731d0ceb4df1a3b57269107e86b6 : SUCCESS in 48m 00s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/14c493ceffa5461496e34d8a230cf7ee : SUCCESS in 46m 51s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/38fa89a8f1e9481ea19c7204393f23ab : SUCCESS in 41m 32s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c42a126806594226a8d753385376dd9f : SUCCESS in 42m 48s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/373ca778d01c42989c27a039cddbf83f : SUCCESS in 46m 16s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/2bdc773019494de8becca486b44a8c08 : SUCCESS in 43m 01s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/73f201d05fca4d81a0edea9db04acf75 : SUCCESS in 40m 13s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e57e594b6cc24e6e97c6c09475ee3c05 : SUCCESS in 38m 03s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48aae6dfb74c47e8aea532e66145edc7 : SUCCESS in 1h 12m 17s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/360655e62d40405aac19f85974829b0d : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2004cc20b6ab96d1afc8c1bb609a2f32ea1b1e9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 20:48:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4f4cf030f7dd9e2daba38b2740fcfae8daf01be5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 20:49:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff325535f46a42a19971486d87252de3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4c337d64f0458681586ff628b8f99b : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c38a98faa574f4982346fa9b16ed54d : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0b8547d7d952f83d6865a935e8e941d7700dd1ff","revisions":{"afcb6a85336f26339e6a05558abfb179b8d674b2":{"kind":"REWORK","_number":1,"created":"2024-05-02 08:07:57.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/27/917927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/27/917927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/27/917927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/27/917927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/27/917927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/27/917927/1"}}},"commit":{"parents":[{"commit":"b47833ca9e781fd3cc80d6bac848acaca1c86c8b","subject":"Merge \"CI: Add a non-voting SLURP upgrade job for Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b47833ca9e781fd3cc80d6bac848acaca1c86c8b"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2024-05-02 08:06:51.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2024-05-02 08:07:52.000000000","tz":600},"subject":"Make boot_interface customisable","message":"Make boot_interface customisable\n\nThis change adds a boot_interface variable to baremtal group_vars to\nprovide the user with a simple way of specifying boot_interface for node\nenrollment.\n\nChange-Id: Ic23dad7a0371c1cc68aa5f870225d6301ed4ed8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/afcb6a85336f26339e6a05558abfb179b8d674b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/afcb6a85336f26339e6a05558abfb179b8d674b2"}]},"branch":"refs/heads/master"},"5b5925006b504e45c0466d9d6fb69879d8b42510":{"kind":"REWORK","_number":2,"created":"2024-05-02 08:26:41.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/27/917927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/27/917927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/27/917927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/27/917927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/27/917927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/27/917927/2"}}},"commit":{"parents":[{"commit":"b47833ca9e781fd3cc80d6bac848acaca1c86c8b","subject":"Merge \"CI: Add a non-voting SLURP upgrade job for Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b47833ca9e781fd3cc80d6bac848acaca1c86c8b"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2024-05-02 08:06:51.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2024-05-02 08:26:29.000000000","tz":600},"subject":"Make boot_interface customisable","message":"Make boot_interface customisable\n\nThis change adds a boot_interface variable to baremtal group_vars to\nprovide the user with a simple way of specifying boot_interface for node\nenrollment.\n\nChange-Id: Ic23dad7a0371c1cc68aa5f870225d6301ed4ed8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5b5925006b504e45c0466d9d6fb69879d8b42510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/5b5925006b504e45c0466d9d6fb69879d8b42510"}]},"branch":"refs/heads/master"},"0b8547d7d952f83d6865a935e8e941d7700dd1ff":{"kind":"REWORK","_number":3,"created":"2024-05-03 06:21:23.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/27/917927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/27/917927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/27/917927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/27/917927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/27/917927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/27/917927/3"}}},"commit":{"parents":[{"commit":"b47833ca9e781fd3cc80d6bac848acaca1c86c8b","subject":"Merge \"CI: Add a non-voting SLURP upgrade job for Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b47833ca9e781fd3cc80d6bac848acaca1c86c8b"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2024-05-02 08:06:51.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2024-05-03 06:21:09.000000000","tz":600},"subject":"Make boot_interface customisable","message":"Make boot_interface customisable\n\nThis change adds a boot_interface variable to baremtal group_vars to\nprovide the user with a simple way of specifying boot_interface for node\nenrollment.\n\nChange-Id: Ic23dad7a0371c1cc68aa5f870225d6301ed4ed8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0b8547d7d952f83d6865a935e8e941d7700dd1ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0b8547d7d952f83d6865a935e8e941d7700dd1ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
