)]}'
{"id":"openstack%2Fbifrost~918029","triplet_id":"openstack%2Fbifrost~master~I55464209f0c876bf42af3c2f6976730c1739a0b6","project":"openstack/bifrost","branch":"master","topic":"update-supported-python","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-06-09 09:47:02.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-06-09 09:47:02.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2024-06-04 07:15:26.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I55464209f0c876bf42af3c2f6976730c1739a0b6","subject":"Update supported Python versions","status":"MERGED","created":"2024-05-03 08:18:07.000000000","updated":"2024-06-09 09:47:54.000000000","submitted":"2024-06-09 09:47:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918029-update-supported-python","meta_rev_id":"53dd8f8795327af09cf61e2bde6262fd3ddf1053","_number":918029,"virtual_id_number":918029,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-09 09:47:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2024-06-09 08:19:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-04 07:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2024-06-09 08:19:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 09:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 14:22:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-05-19 15:28:38.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2024-06-09 08:19:54.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"}],"messages":[{"id":"ff692fd5e5c0d8bac73813d421bb29df4e3d5470","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-05-03 08:18:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"817d4f2f3ae6720c74ae7062186bf4df9041d4e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 09:51:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3e23290c524bc597b6035f11a301fe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a9b8b3f0a1f4213b61872b4a3a1470d : SUCCESS in 1m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a57792d8ccf94c9fbdf536272d126c11 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddfb4cea755c4dd8939e072dd7eceb07 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82a0ffc3ac094591860d270178f9517d : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22fc0906be7943408453dc607983bb55 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a40b63d13b643e3a3af50b67ff42ec4 : SUCCESS in 5m 24s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/738570cb40d3430796b59ef7b645bf6f : SUCCESS in 8m 25s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/1c0e9e1fd0534928ad60286ab4082f8b : SUCCESS in 32m 41s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a20a3d6cfd6348f08fef791497a0f6f5 : SUCCESS in 36m 41s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9c733eecd62410185f47a173785cb2d : SUCCESS in 41m 06s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/951c82e173ee4f2886401d0687fb3def : SUCCESS in 43m 27s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/e7c12f3a415b4564a7a2840982f23224 : SUCCESS in 43m 43s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/edeadf3ff52c4808b3f1009265585476 : SUCCESS in 43m 24s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f74bee71454455bbd46a8ed932c5e27 : SUCCESS in 36m 22s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/08221a618929437593fe9622e79817a2 : SUCCESS in 42m 07s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/3a7af87789804be3814a293350899f79 : SUCCESS in 36m 30s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/f4f63f20532c49e19cd2a9f8c785a603 : SUCCESS in 41m 25s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00cff6c989b24a5abbc1c74a74ef9903 : SUCCESS in 40m 29s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76d662a1d7854a1da7d3bfaabd8303cb : SUCCESS in 1h 08m 10s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/3917062efa324d81964f473a90488dbd : SUCCESS in 1h 28m 28s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e6c2053c8c84ad48926739e827a957e : SUCCESS in 59m 29s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/bef55fd08b954c15916de73a4368e4f5 : SUCCESS in 32m 22s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c503cb2c6c324bac9039d36388c77fed : SUCCESS in 41m 09s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/a9fb4358a7554087907677abfb6a4b26 : SUCCESS in 41m 05s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bc0e7ecc8e54a39a19cb2eb3062e266 : SUCCESS in 34m 26s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/717380cc569e49d59a07c8add5074823 : SUCCESS in 46m 24s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a25bf5419d0a43828d05683cfef6d04e : SUCCESS in 46m 07s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/950154fce6604b0eaafc4b3e46aa3a7f : SUCCESS in 47m 12s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/47c772029c0a4dd1863c92e5b0b3942d : SUCCESS in 41m 04s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/c8f14354666d450783de9c611b2d2c9d : SUCCESS in 20m 25s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/0cf6d661708c437b8e5415925d49edea : FAILURE in 2m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c627ff8abc899d630f5a588d777dae297fe8e6a5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-05-13 14:22:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8e24725d0b1f18745b3237aa8b1865ee72e6a76","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-05-19 15:28:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7089afc501ee79c141409321ce478cbed416a47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-05-29 09:37:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"94328a8ca88ef44c5f41f2b9ea06b70128db9139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 10:57:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30fc5909483c4253ab1353fdac68c14d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bae3be2dce46488fb76f1371e2c59544 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f395b5454f2041ab8d4a6a673706d7a5 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63bc50e7ad92449dacc6a01e6415f2e8 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f288234385804797a36e654d44aaf121 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98f78fde6ac14399b70da656120aea7d : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3b2fdc98e74c82b09c1d18b31534b3 : SUCCESS in 5m 12s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/0d967920364a4a329749c380357b9c1a : SUCCESS in 7m 41s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/c53d3076531e45c9881d55c956dfecb0 : SUCCESS in 45m 49s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0b980c430c24e3c937c2b829202d4b7 : SUCCESS in 44m 56s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ae1f5ed68374619b1228016cab5241a : SUCCESS in 41m 58s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/0b2ed9cff4c645c297814086da507108 : SUCCESS in 47m 18s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/ff98a861e75e4dcea50e02cc1f5b70e2 : SUCCESS in 38m 46s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/280f5db9b9754c2e9673e734c153e4d7 : SUCCESS in 42m 55s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dafa7578564b41238ea6929870767a79 : SUCCESS in 44m 23s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/b5e1cb66dc4d4742a2a5f852f6acee91 : SUCCESS in 48m 22s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/fb9b4c1403414c60a7aee072b3e6e5a7 : SUCCESS in 41m 55s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/53716c7da62c4d04981249419fa3da3e : SUCCESS in 42m 14s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/909b54835768454bb5433ae899eeb93e : SUCCESS in 37m 43s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fd590f86f314f06b115c93348ce961f : SUCCESS in 1h 03m 18s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/2e8d667ac54a4dc2a2554bff6029a3eb : SUCCESS in 1h 17m 09s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64d4c0b6b1e940c7bbc8175d4f42e46b : SUCCESS in 1h 12m 06s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/a5aeb28551584301a8aa10c6400a650c : SUCCESS in 38m 01s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc01eed1dcc744f6ba2854c307a3fc74 : SUCCESS in 39m 22s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/5627f1c0c60f47f4a333d74bb59be860 : SUCCESS in 42m 26s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bff2f059b95411d9ebc3fe442447889 : SUCCESS in 40m 22s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/b24627cb71c64f78a97d9ff10b612902 : SUCCESS in 40m 24s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22fabf46ec854caaa9aa886e7c696bbd : SUCCESS in 48m 16s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/28463095a0114d28815caa613f027cb1 : SUCCESS in 45m 16s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/f78c3b300e794efea0ce14c3b9cbc837 : SUCCESS in 44m 26s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/e927788f30ae4234b5e6bfc3245256d1 : SUCCESS in 23m 09s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/559e047744b54b5384f5a03753a15e1a : FAILURE in 2m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87439f9008cac294dd7c58d5c51e21a60ae04909","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-06-04 07:15:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6b4be355df6f2acd206f5bf9b004d11931c20a21","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2024-06-09 08:19:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8ab546a539f0ffa2e1ffb0b5b410dbc64dbd4b00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 08:20:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"733b1dc5e35c16d305c4f59827f7a8034e4afc06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 09:47:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5ba055cc6714817986d0f3a768efd6d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/81e28bb982e348b6b2cda5c3b7ea29ad : SUCCESS in 1m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cbabdf299f844af89423fa0108ced89 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/090037db9e1a493d96f13d17776da8d1 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/835ef35875454960a5551f79488fd4a7 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c9a4368da94a7d913dd16f51c93d33 : SUCCESS in 4m 47s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/1b304ba3fc9f4cb6878e090b0178c5d9 : SUCCESS in 7m 48s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/b012a238ef7b4bf999373a538e2cc44f : SUCCESS in 34m 42s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfa67adddda542e596390519e9653f9a : SUCCESS in 45m 10s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f260126e159404ebd404c6d76ecf679 : SUCCESS in 43m 29s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/325b8dfac52d465e9934d774c5e11eca : SUCCESS in 46m 22s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/d1eb18b3676d4112a6d6857fefff2128 : SUCCESS in 43m 54s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/3edb3a33eedd46d1b73722e2c37cd376 : SUCCESS in 42m 28s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc43672ffe2c42cdb61cf8456475c22f : SUCCESS in 42m 56s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/b01a6f7a19c74814b06057220f62e767 : SUCCESS in 45m 52s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/e84696eb334b44658ccaa79d0b3d9c48 : SUCCESS in 42m 52s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/6b63e59054c6447ab98b5584e1b103a3 : SUCCESS in 43m 06s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/683c61cb6ec14766abed3ac9cee3d0d3 : SUCCESS in 40m 11s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78486488daff4309be1c9a81766e9e60 : SUCCESS in 1h 21m 43s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/c679ffc77c2d4edc8ab29023c86d1d61 : SUCCESS in 1h 17m 03s","accounts_in_message":[],"_revision_number":2},{"id":"263fbe68842b7da8ac1663753686d6d207a4090e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 09:47:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"53dd8f8795327af09cf61e2bde6262fd3ddf1053","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 09:47:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a02a519d600410db95c90f61c806edf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adafd2cabd6b4e6c8cf7e9a3f285f365 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"797d622f3d6b6c462de77cfd89e32e566def8177","revisions":{"95d7f3ad9c26de32a7ccd3f75c5c9919aa062ad7":{"kind":"REWORK","_number":1,"created":"2024-05-03 08:18:07.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/29/918029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/29/918029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/29/918029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/29/918029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/29/918029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/29/918029/1"}}},"commit":{"parents":[{"commit":"b47833ca9e781fd3cc80d6bac848acaca1c86c8b","subject":"Merge \"CI: Add a non-voting SLURP upgrade job for Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b47833ca9e781fd3cc80d6bac848acaca1c86c8b"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-05-03 08:17:39.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-05-03 08:17:39.000000000","tz":120},"subject":"Update supported Python versions","message":"Update supported Python versions\n\nMinimum supported Python version is now 3.8\n\nChange-Id: I55464209f0c876bf42af3c2f6976730c1739a0b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/95d7f3ad9c26de32a7ccd3f75c5c9919aa062ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/95d7f3ad9c26de32a7ccd3f75c5c9919aa062ad7"}]},"branch":"refs/heads/master"},"797d622f3d6b6c462de77cfd89e32e566def8177":{"kind":"REWORK","_number":2,"created":"2024-05-29 09:37:51.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/29/918029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/29/918029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/29/918029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/29/918029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/29/918029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/29/918029/2"}}},"commit":{"parents":[{"commit":"e0a5a7216d74821aee8cc32a8767c790ed731483","subject":"Consolidate centos/fedora/redhat required_defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e0a5a7216d74821aee8cc32a8767c790ed731483"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-05-03 08:17:39.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-05-29 09:37:47.000000000","tz":120},"subject":"Update supported Python versions","message":"Update supported Python versions\n\nMinimum supported Python version is now 3.8\n\nChange-Id: I55464209f0c876bf42af3c2f6976730c1739a0b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/797d622f3d6b6c462de77cfd89e32e566def8177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/797d622f3d6b6c462de77cfd89e32e566def8177"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
