)]}'
{"id":"openstack%2Fbifrost~918825","triplet_id":"openstack%2Fbifrost~master~I6cfa9af05655b496e58e211bf45475e7f71618a8","project":"openstack/bifrost","branch":"master","topic":"bump_sphinx","hashtags":[],"change_id":"I6cfa9af05655b496e58e211bf45475e7f71618a8","subject":"[DNM] Testing docs bump with new Sphinx","status":"ABANDONED","created":"2024-05-09 19:22:05.000000000","updated":"2024-05-09 23:08:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"49a47324cb7d61abdc664e81cdfce90645e37667","_number":918825,"virtual_id_number":918825,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-09 21:24:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-05-09 21:24:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45793603cf2d6c876dd2abc4f388550db2f07639","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-09 19:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20e856774c7f6fcd85fcc8bf8b3bd5e9b2353abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 21:24:41.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2d1f7f4a685465fa00cd5d6da41e523\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/14f63cbc0b3441d2b34b809ef7d9d34a : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4641c80d120475e91ce63fb49975a55 : FAILURE in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/893d47b5cd9245d1907611ca45436aa1 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68238890ccb048b6bbe057ce2adbd237 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6538e43a0b614fc39fc6a1f4bbd7f1c2 : SUCCESS in 5m 43s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/da9718cac7e0428c97b54fad772e6ed7 : SUCCESS in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed3f1d8b287430f9278135fa83d5d75 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f97591380da641389859ab2232a28761 : SUCCESS in 3m 10s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/018549f5cede48f9ae6bb5a35d1b4fad : SUCCESS in 8m 22s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/4e250b8c3a934ef5a6162bde4e0a329c : SUCCESS in 34m 26s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7943134266bb4db888c7e97f8db14b91 : SUCCESS in 50m 53s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3492e618ec004e23a558caba2af31167 : SUCCESS in 41m 15s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/7b89eb697c884c5bb88c0f6c7fdad71f : SUCCESS in 54m 00s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/bfa59d3fbc9f4c559aa2db5e745f1db6 : SUCCESS in 49m 43s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/1a8545af4606449a805bff2c373d2dc0 : SUCCESS in 40m 43s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08e5e15db65c4988a82a3c4e08ddccbb : SUCCESS in 44m 13s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/b6fad794e1e542a4ac85fd3fad32421f : SUCCESS in 50m 51s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/6abaa1a1d86349679aaac6c4f1aad142 : SUCCESS in 45m 35s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/24a8ec99091d40bda8ba7bfc07821dbf : SUCCESS in 41m 16s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02afb2b44d034ae1b2c960f7fc4022b4 : SUCCESS in 31m 53s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dba9b4fb1800496e949f63ae93752748 : SUCCESS in 1h 33m 26s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/b9cd7395450e4e76a5966d3677cd1fd8 : SUCCESS in 1h 30m 11s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26208396215f4195bf9ba4f3b5982dd1 : SUCCESS in 1h 27m 58s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/3956fdd7b9ec4010ac761b781f769a96 : SUCCESS in 37m 32s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3bcdc44f2c54300927864c0cf7a86fd : SUCCESS in 39m 18s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/ab90c6d315f040a79473965625b17bfc : SUCCESS in 41m 20s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2496daed492b4f099418fb3084dd2eb0 : SUCCESS in 48m 21s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/afcd425d763a4752b171623d0809de2d : SUCCESS in 51m 39s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56a8f2ba255f4f88ac6a8a742b368f39 : SUCCESS in 50m 30s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/97078b9a147b4f0698da487c16489545 : SUCCESS in 49m 19s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/f18434af5ab64465bff9606b3f5114a3 : SUCCESS in 48m 53s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/f841756ddb044037a6db8119ea1a3519 : SUCCESS in 28m 02s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/dfad359bed4a4a6ba9fe7589ac7b69f0 : FAILURE in 3m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49a47324cb7d61abdc664e81cdfce90645e37667","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-09 23:08:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"add053ea14464bbfb99df2588f20debd320320b6","revisions":{"add053ea14464bbfb99df2588f20debd320320b6":{"kind":"REWORK","_number":1,"created":"2024-05-09 19:22:05.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/25/918825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/25/918825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/25/918825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/25/918825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/25/918825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/25/918825/1"}}},"commit":{"parents":[{"commit":"4c8c2406f77edef9caf0f6dd858ee54920458b7a","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/4c8c2406f77edef9caf0f6dd858ee54920458b7a"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-09 19:22:03.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-09 19:22:03.000000000","tz":-300},"subject":"[DNM] Testing docs bump with new Sphinx","message":"[DNM] Testing docs bump with new Sphinx\n\nSphinx updates can be scary.  This change leverages a constraints bump to see what breaks.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/916565\nChange-Id: I6cfa9af05655b496e58e211bf45475e7f71618a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/add053ea14464bbfb99df2588f20debd320320b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/add053ea14464bbfb99df2588f20debd320320b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
