)]}'
{"id":"openstack%2Fbifrost~922789","triplet_id":"openstack%2Fbifrost~master~I5df83ec54ecd7b8b363900d7ec7ada1ed9de0f5a","project":"openstack/bifrost","branch":"master","topic":"remove-cs8-bits","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-07-04 18:22:30.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2024-06-27 12:42:23.000000000","reason":"\u003cGERRIT_ACCOUNT_10239\u003e replied on the change","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"hashtags":["ironic-week-prio"],"change_id":"I5df83ec54ecd7b8b363900d7ec7ada1ed9de0f5a","subject":"Remove CentOS Stream 8 leftovers","status":"MERGED","created":"2024-06-26 08:31:19.000000000","updated":"2024-07-04 18:23:38.000000000","submitted":"2024-07-04 18:22:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"922789-remove-cs8-bits","meta_rev_id":"9965f802e74fc193f78ef1a078a484c337f3c80e","_number":922789,"virtual_id_number":922789,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 18:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2024-07-04 16:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-27 12:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-07-04 16:45:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-26 08:41:05.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2024-06-26 10:12:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 16:45:50.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"f78974c16f06676ac8a65ea8d82818b65996c735","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-06-26 08:31:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbe0e694b8a1d4c53ec2ac2597666fca0eb4c85b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-06-26 08:41:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dc4d4ddcafe0eb9e27dadcc0f35e3b8b5a01adde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 10:12:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b99d607bb1d74040a1a0b3020943e5d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/779aabc17acc426083f8f3f488ed1b61 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5bd246a8bf74ec49246e86fcb9ff06a : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8551dda4d4c448d6a31aa5d989b6a794 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5275c2421dd442cd9de9f17898f10e1a : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe818c2b2874d9080a9433da78ea777 : SUCCESS in 5m 02s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/cb64276778ee42e181300d9378a269d3 : SUCCESS in 6m 50s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/e4bb3b121f2643e58ed78dc7e5baa79b : SUCCESS in 43m 15s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c699b20ca787404c80a6c8ae6bbcc8f9 : SUCCESS in 35m 59s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4685cc09d93645749e3bef511d0359ce : SUCCESS in 40m 12s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/b28847fd37d840918488ac2d83884cfd : SUCCESS in 44m 04s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/eeea15ea56b342f9a598d2d79e42e5bc : SUCCESS in 41m 10s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/b7a3bc16ea8e402db196752b63871943 : SUCCESS in 40m 04s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba9560a112dd454caa32f1f1ff8ab160 : SUCCESS in 38m 50s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/bbcaeac96b6141da9753c02b25b9de4e : SUCCESS in 45m 07s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/43148b033cb345f59fe1bb9313dbf009 : SUCCESS in 40m 37s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/3688c0622e9e4163b8be314a601b4f3f : FAILURE in 47m 25s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e585ca9f56f46a287ae3976b25cce8e : SUCCESS in 40m 10s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b74f49994fb4577b1449619e7fcb507 : SUCCESS in 1h 17m 51s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/301455bc63d448d09fa4642c6ad6726f : SUCCESS in 1h 35m 10s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6464ff65abfa4e60829f2a6923e756d3 : SUCCESS in 1h 01m 57s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/d741103e966e4ed5801ded74ec1a6834 : FAILURE in 53m 38s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f212228feb54c6e921e637e646001f4 : SUCCESS in 40m 42s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/9a50d6f404e449cca2fec525abb602a2 : SUCCESS in 44m 08s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/790e08c6ef90485a9d68b7801bc1790d : SUCCESS in 34m 37s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/f1476affa01746f79a6c7d7ac8810adf : SUCCESS in 37m 00s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f20c9e9b4cf4ad28aedb447c95d9eab : SUCCESS in 45m 58s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/050f1304e5684cca8521f959e8373379 : SUCCESS in 38m 34s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/2f161af197414491a091f14094a51e4a : SUCCESS in 43m 50s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/5e1dc18c6f3b4ead9d44a279bd00fe27 : SUCCESS in 24m 12s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/2e761bc3f939469d9dc5f3eb855360a1 : SUCCESS in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"996cc7b25ed7ed5a88d7dd3bc68d86f8a9a83be2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-06-26 13:54:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"388037c1722347bdf98c0576ec13cd4f7d3225e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-06-26 13:56:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3f9cb128ede7675db62d1bd582da23aeb17912b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 15:28:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa8a149d69c04bbdbae5e382d8727e8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6669af090244d5a4cd4065bacdd4ef : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d394a2d60ad44298025c5774697c363 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f12bd071ae634fb2a8638a87ef1fd598 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c200642b10d9413fb76b8a393daef52c : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598dbb01bf1948c29e50ee1584384e4c : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f9b83b2552544fb88e534062ce4b637 : SUCCESS in 2m 49s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/0565ca965bfa4e6b8337969cc221575e : SUCCESS in 7m 36s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/ba5dab75219b43eca6614b7e97193608 : SUCCESS in 45m 20s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d02e39c0ccb04b919c7b2545b8e61837 : SUCCESS in 42m 48s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a62632ef34541fa9f2425fc5ed83715 : SUCCESS in 44m 31s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/01214951c0c24de5844bb386d1b33653 : SUCCESS in 47m 34s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/8d8da1db6a7e48dfbba14e846d9f1099 : SUCCESS in 40m 12s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/63fced82842549e8b65322dbc4bd4de4 : SUCCESS in 39m 10s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03cc90fbf9964da6bef039ad74e5a5c4 : SUCCESS in 40m 10s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/bc323acfe5cd4fe28da341e6ab8a8661 : SUCCESS in 45m 49s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/17948ad9b2a74c97b78e1ee076333f5c : SUCCESS in 43m 24s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/3672977d8382413ab3a2113985f29ed4 : SUCCESS in 40m 30s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b3d6a440a4343c3a6101a7d045cafa3 : SUCCESS in 36m 22s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4dee629884bb4c119a4382c283006007 : SUCCESS in 1h 15m 54s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/b6a18f6c96a8455eb08d03535c124c14 : SUCCESS in 1h 29m 47s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a8b6aab9e6e4edb970a92909f5e01a7 : SUCCESS in 1h 06m 33s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/aeafb064ee32450bb4fe419f0ae0e4ea : FAILURE in 53m 37s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/248004ad36404f03947f0791b6ebaab7 : SUCCESS in 39m 10s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/c601a0f2497a409b8d3dc594c0dfb2db : SUCCESS in 43m 10s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f230536898e947319b930cc2f23c7806 : SUCCESS in 36m 22s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/4d1aba6e73314ca3b94801680cf5cf2b : SUCCESS in 37m 58s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b815f7d7fb34d11a0180311dea077c2 : SUCCESS in 41m 37s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/e152336e3b644958a5f6ec79f0a02050 : SUCCESS in 41m 04s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/a008e698c3fd49378f0f078ea03a964d : SUCCESS in 41m 25s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/e2388861ea05472f83bef01ad62c07df : SUCCESS in 22m 12s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/e33656fe68844fb793ce1fb9bad6774a : SUCCESS in 2m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b4572f83b2f80ce90149c755cd338a31a962dca","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-06-27 12:33:47.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"37f9ab4f7c9db387a4fd36b3e6058e7a428ae8b2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-06-27 12:42:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ec7ec27dc3baee8f922005c99a47a0d79767cdba","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-04 16:45:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"90cd988ae85b53588aaa3e2408d41516994dedc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 16:46:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5376aa230dd6628df77abd3473689456d0f17fa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 18:22:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43598f57ac2d420d96aecb4e93a6f0c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263e3acc2cf645a79ad9e7c1186bdb8e : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b783eec85ba94e1ab2cb3e51f1b8bfe7 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44c066232a20491dba54a0b217fefd9f : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f9cbc850cef45708e2ceb38d4bc2a54 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7df30317ab494a14913ba04ab8659662 : SUCCESS in 2m 44s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/3c426982fa92431aa448e2ed1bbb0f46 : SUCCESS in 8m 32s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/81b7cbc994e44a35a11bc8d2d74c2961 : SUCCESS in 47m 53s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf5e696faf7b4d898f9e8d3705d8ed27 : SUCCESS in 44m 16s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/764924529d474edc9f47ae2e86f8c622 : SUCCESS in 46m 37s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/eb6cfd5d12ea4f6d85324c4211cab4ed : SUCCESS in 50m 48s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f762067615d4b33953ace7ebd652645 : SUCCESS in 44m 37s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/502da63fa01640bf8ca5cfb0e7dab684 : SUCCESS in 43m 59s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5aca0c652584443eb5d58a867bddbc06 : SUCCESS in 44m 50s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/365effc7add44e2dbb3abe881a72b444 : SUCCESS in 40m 06s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/88677ccabf68415fbd0c46a318549e29 : SUCCESS in 43m 26s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/3f9533574a8b40fd94f80743104c0ad0 : SUCCESS in 39m 39s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3408eab17ce94fb39b754736034dc352 : SUCCESS in 40m 36s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/331c3e245c634d7aa9223b893df52d99 : SUCCESS in 1h 19m 44s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/d1cdb390de704c71b7803f8126751792 : SUCCESS in 1h 30m 18s","accounts_in_message":[],"_revision_number":2},{"id":"5b3a39b07fff9b356a790fc6bea3948be50ac72c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 18:22:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9965f802e74fc193f78ef1a078a484c337f3c80e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 18:23:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0c6672c8d2242eda2dddae6d6520156\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36ca29c764445ca9b23cdbd1d741cb6 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/339aa35abaf84e53968bda67c0804665 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cebecff3cc6e1458be03c2f0c4f8bbfca81a15b2","revisions":{"2574f28ef659a5ae11b525e48a4513f4b31b7642":{"kind":"REWORK","_number":1,"created":"2024-06-26 08:31:19.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/89/922789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/89/922789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/89/922789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/89/922789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/89/922789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/89/922789/1"}}},"commit":{"parents":[{"commit":"cd947080ad558f3da8a9310814362ee6023fa156","subject":"Merge \"Fix codespell reported errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cd947080ad558f3da8a9310814362ee6023fa156"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-06-26 08:30:45.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-06-26 08:30:45.000000000","tz":120},"subject":"Remove CentOS Stream 8 leftovers","message":"Remove CentOS Stream 8 leftovers\n\nAnd yes, even some CentOS 7 ones\n\nChange-Id: I5df83ec54ecd7b8b363900d7ec7ada1ed9de0f5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2574f28ef659a5ae11b525e48a4513f4b31b7642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2574f28ef659a5ae11b525e48a4513f4b31b7642"}]},"branch":"refs/heads/master"},"cebecff3cc6e1458be03c2f0c4f8bbfca81a15b2":{"kind":"REWORK","_number":2,"created":"2024-06-26 13:56:29.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/89/922789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/89/922789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/89/922789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/89/922789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/89/922789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/89/922789/2"}}},"commit":{"parents":[{"commit":"cd947080ad558f3da8a9310814362ee6023fa156","subject":"Merge \"Fix codespell reported errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cd947080ad558f3da8a9310814362ee6023fa156"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-06-26 08:30:45.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-06-26 13:56:20.000000000","tz":120},"subject":"Remove CentOS Stream 8 leftovers","message":"Remove CentOS Stream 8 leftovers\n\nAnd yes, even some CentOS 7 ones\n\nChange-Id: I5df83ec54ecd7b8b363900d7ec7ada1ed9de0f5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cebecff3cc6e1458be03c2f0c4f8bbfca81a15b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cebecff3cc6e1458be03c2f0c4f8bbfca81a15b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
