)]}'
{"id":"openstack%2Fbifrost~936797","triplet_id":"openstack%2Fbifrost~master~I47413eab8c5132ab3201b4c7197ec15c17d18e12","project":"openstack/bifrost","branch":"master","topic":"rhelatives","attention_set":{},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2024-12-06 14:48:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47413eab8c5132ab3201b4c7197ec15c17d18e12","subject":"Use python version to set DEFAULT_PIP_ANSIBLE","status":"MERGED","created":"2024-12-01 12:32:42.000000000","updated":"2024-12-06 14:49:33.000000000","submitted":"2024-12-06 14:48:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936797-rhelatives","meta_rev_id":"af41a4e1650d680a42a6821080e0f125d08bff21","_number":936797,"virtual_id_number":936797,"owner":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-06 14:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2024-12-02 13:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-06 13:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-06 13:41:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-01 14:05:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 13:52:33.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-12-06 13:41:10.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"210ca59dac9ee831f390afdcd3320b00aaec9c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2024-12-01 12:32:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b04e406cf1e4da25a719950e112f8aa3e9a8fb8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-01 14:05:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c2f39ec338d4868911923ff1a2874e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/202514fec55e4b4ea9c3ea1ba4bfdc90 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca2c7d91e78548ed9a0adbaa249de726 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d453c38d371b4b45baf5d47c57d0b64e : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be227db8b6a84642a15fbc4a59154930 : SUCCESS in 3m 38s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/4600b644cfa040aaa91c6a01b49d185a : SUCCESS in 7m 09s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/e254a3f4a60245ff908eda155276c365 : SUCCESS in 35m 03s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a4dc4a29371416c923eaf8d4baf88f1 : SUCCESS in 29m 55s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d75f865c64e1440381c952536156c038 : SUCCESS in 31m 38s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd6180ab95da4dcbac4af15c3666d4b2 : SUCCESS in 30m 58s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfbb0f9ec7554e5eb65e88eee5ae8a8a : SUCCESS in 42m 30s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/7f3597de172948a186a17e0fcf017181 : SUCCESS in 47m 30s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/ab2122686f234d8bbcde0da934bb6993 : SUCCESS in 39m 38s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/a46dc26b6fd24c679af1df062e742afc : SUCCESS in 37m 38s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60d7311fc77346e7b50108435c999a8b : SUCCESS in 30m 10s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a93cda15680448fbbbd0796a7f01313f : SUCCESS in 29m 36s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/db759d86c35b4acf9e3d52ed51b6185e : SUCCESS in 30m 36s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/ed76622065484dc29396a98f4583592b : SUCCESS in 20m 04s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/c50f60820be2443f8fe8f9c633259f8c : SUCCESS in 31m 41s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65f7b0b2f5334440b359058d5885a503 : SUCCESS in 41m 41s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d0ec1926e8c4bbfb258f7e970e4a05d : SUCCESS in 1h 10m 00s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/e6223bcee6fd4f699c47961f73a2efe7 : SUCCESS in 1h 13m 09s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64a9d1ad54fa4153b968d8e00f3833e7 : FAILURE in 1m 21s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/185abe3285974b28a813eb8be0b29f8b : FAILURE in 45m 47s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ab6ed2a062e4ab294595015578fa84a : SUCCESS in 30m 43s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/975fe44d7ade4836824a0e40ae55ca05 : SUCCESS in 33m 30s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35c09701483743918a34923d1cdd170f : SUCCESS in 18m 40s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/7b4dcf773dde4d4d9856aaf0bd958d3b : SUCCESS in 34m 45s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11faf1afcba542cfa73923386044f276 : SUCCESS in 32m 37s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/d9c6ef5a92e24591a30728c7ab9ff9e6 : SUCCESS in 30m 47s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/1fb132577b8b4a25a6cfde492d4f1066 : SUCCESS in 41m 09s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/aff9fcc0b449419288846b5031643a30 : SUCCESS in 20m 06s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/98009ef78ff442f5868eb6addd37f811 : SUCCESS in 2m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de6b5685e42bee5ce0376b08d8f5f41bc376f54b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-12-02 13:52:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f23d7087d0f5f793e680dfc22a70981428023e5e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-12-06 13:41:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62f8d10963985efaf7c92ce1e7caee6cfecdb6e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 13:41:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"764e7d1681259491b3568cb51651a6059dc0f9d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 14:48:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64d68c94ac1a470cb29be9a908d0e686\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/315beb8fac514eac81f79ae92cd22f23 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54228ea4193d420db4c5aeecb825628e : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c54bd99a0e5549d2a898c6623688034d : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2a9e18607b4d489f7b9fc1e4d4000a : SUCCESS in 5m 45s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/3d764d07213a483b80c933f08e5fe432 : SUCCESS in 9m 16s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/6d5e0ced40bf499fbfdbd0aa97ba6b6d : SUCCESS in 43m 54s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39a9ec6b8953416f9a19333a53a18704 : SUCCESS in 40m 20s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3c225268c4347daa8f0233bd1662337 : SUCCESS in 42m 11s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d106c9503aa47c9912fa26c53b2df81 : SUCCESS in 45m 07s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4011b9ab43ec414d871622655e0695c6 : SUCCESS in 40m 22s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/7f65c25286324849826e61b9fffec1c7 : SUCCESS in 34m 52s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/1a81bfd2c7604f46a2ba1a3ef325c784 : SUCCESS in 37m 45s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/6b0d19dde5214cf884b788fd3c46235f : SUCCESS in 40m 17s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d259e5393fae4bc4a34958d7eea3c286 : SUCCESS in 22m 09s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9118c3d26914e30ba67305b9ca04e3b : SUCCESS in 42m 44s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/ebc0e5d4dc724692bd5ea6cffe035abb : SUCCESS in 37m 56s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/1f87e5bea28b4698b254100c1a2cc73f : SUCCESS in 46m 13s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/9c9f69d6ddf84ebc82eb0a4a602cb3fc : SUCCESS in 43m 16s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6995e01c8d2044ebb78b2ffb8e206e93 : SUCCESS in 38m 50s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8c6429213f24b618129958235e736f5 : SUCCESS in 44m 31s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/a52cbcc1aea04921b2b484980d2c99d5 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fda9c9d06d67469f062ce255dc674c4580be35a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 14:48:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"af41a4e1650d680a42a6821080e0f125d08bff21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 14:49:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7f8b84435d45d5b8c8727f8d1396dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94344549bd404dcf8786e4e00b91a80e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76f82acff814a79ca36a6034ba23bf4086dc9449","revisions":{"76f82acff814a79ca36a6034ba23bf4086dc9449":{"kind":"REWORK","_number":1,"created":"2024-12-01 12:32:42.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/97/936797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/97/936797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/936797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/936797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/97/936797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/97/936797/1"}}},"commit":{"parents":[{"commit":"2d6ccc2b325e43c5c68b43ce2e9c3ee5e3265388","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2d6ccc2b325e43c5c68b43ce2e9c3ee5e3265388"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2024-12-01 12:03:46.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2024-12-01 12:19:50.000000000","tz":60},"subject":"Use python version to set DEFAULT_PIP_ANSIBLE","message":"Use python version to set DEFAULT_PIP_ANSIBLE\n\nInstead of assuming the python version based on the distro, we\ncheck the python version. (Check borrowed from bifrost-cli)\n\nThis patch defaults to the newer ansible version, and if python\nis old, we override it with the old ansible package.\n\nThis lets us install on the RHELatives like Alma and Rocky, which\nthe current check is preventing.\n\nChange-Id: I47413eab8c5132ab3201b4c7197ec15c17d18e12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/76f82acff814a79ca36a6034ba23bf4086dc9449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/76f82acff814a79ca36a6034ba23bf4086dc9449"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
