)]}'
{"id":"openstack%2Fbifrost~940505","triplet_id":"openstack%2Fbifrost~master~I3e830e1c6caa85437c4aa6fbc95e5d384ee9562d","project":"openstack/bifrost","branch":"master","topic":"deploy-dynamic-uuid-to-id-bugfix","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2025-02-03 16:34:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e830e1c6caa85437c4aa6fbc95e5d384ee9562d","subject":"Update `uuid` to `id` in node_info to match module change","status":"MERGED","created":"2025-01-31 10:19:59.000000000","updated":"2025-02-03 16:35:30.000000000","submitted":"2025-02-03 16:34:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940505-deploy-dynamic-uuid-to-id-bugfix","meta_rev_id":"0cbd4c73db1c7ca505f734b71795950cd5338404","_number":940505,"virtual_id_number":940505,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-03 16:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2025-02-03 14:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2025-02-03 11:56:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2025-02-03 14:44:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-31 10:28:48.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-01-31 12:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-03 14:44:08.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"ba567a6f82cb7483b68daccf99400f775920fe85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-01-31 10:19:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce86a2cc10ec2093cbb20e3893a00a01800655b0","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-01-31 10:28:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0282b33ef02da02c38943e8f346c5320fd1fc022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 12:12:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34e61ad1618940f9b7620fba179fe141\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64597179c6724b54b1b0206156789d9d : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e8d02ba4aa74c8fbad105fdba68960f : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eca6d2e3a63d4765836de4940c022ec6 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ec9712a68014256a0fe5dcedfd6419d : SUCCESS in 5m 48s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/26dc4fcc32df45fcae2034685994c370 : FAILURE in 4m 28s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/0312278f6ca04c1dbcf28183e974ad84 : SUCCESS in 43m 26s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/570a0790851249f6ba504871147002e2 : SUCCESS in 23m 25s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd642c40b60b4b8f8eb3a5097a0a7b93 : SUCCESS in 40m 33s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca51da8f69de43ff9fd7f235aa172bee : SUCCESS in 42m 13s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b902bb2a0914b24bb087899030279ff : SUCCESS in 38m 32s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/7b1317d94fe441dfbe31fb5cb75a7358 : SUCCESS in 44m 03s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/c9b336fcf13f483fb078a79c1e7fd3d7 : SUCCESS in 41m 32s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/812097b246e14c77acfb6cb6dbe54f24 : SUCCESS in 35m 48s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4b9d342ec7d45fba3b4abe7295cc41c : SUCCESS in 43m 05s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e1ba0c159434ae0bfa23edf29320d6f : SUCCESS in 41m 38s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/bc2860da253d45c2aba4ecf7c7433409 : SUCCESS in 44m 35s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/46ecd9748170483aa520a16d1faa70c7 : SUCCESS in 54m 44s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/47e17f36b0ba4569af94eed3813f7b3c : SUCCESS in 54m 29s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80979b2739ee4b5885ba54d3808366f7 : SUCCESS in 54m 26s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46ff9e14679d4777aed62ecfc02a68e2 : SUCCESS in 1h 29m 14s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/ba5f1dbbcbf240c1be0825916cd935b2 : SUCCESS in 1h 45m 47s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3380a9fbc3243f1ae31be161a575fbf : FAILURE in 5m 12s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/5ab81bc43727404ea6ffcdb64b2e6d4d : TIMED_OUT in 1h 00m 45s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f835ead33cde401399aee49591a74ea1 : SUCCESS in 39m 19s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/6472fe7021524417a379debfe311a018 : SUCCESS in 43m 02s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62211b5a1db24725804664d72a212f31 : SUCCESS in 38m 34s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/b9b6ba5b24264bfeb864d1f0cb961c50 : SUCCESS in 43m 18s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/118c191c31db4396bfba26d8bc27b030 : SUCCESS in 45m 57s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/5457aba9cd614615b0799a355ef864cc : SUCCESS in 38m 16s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/1dc0f6bdd1734aed9b1214f209cc1d91 : TIMED_OUT in 1h 01m 03s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/729e2f8cab284eabbd1fa9a932906a6a : SUCCESS in 1h 04m 01s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/b0ce80fdf13246db820ad2ac4f33cc20 : SUCCESS in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df25a926f822052bde410fe7e94db391bed62670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-02-03 11:56:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8d4ab782ac10f45e887bdc43cdd5372b0d608b30","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-02-03 11:56:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac6a27ea5deb3d6737e396242bbd1bf5ab8be753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 13:28:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae11bf9c204d43eaa0985171234b43ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a5123165c847c99eb9fa99cba9012a : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25405274e96948eeb269c117ab5c6c9e : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25aea365f35741d19155fb6e64b8fbf0 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a5b04f56ef498c9a73cae5ad8ac192 : SUCCESS in 3m 37s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/9be3eaab525a45e1a78577be688ce040 : SUCCESS in 7m 22s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/ac379d350aa74ed08dc7cab406c528ef : SUCCESS in 25m 43s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1069ec7d82b4db0a64d39a495a1ad44 : SUCCESS in 38m 00s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/976f71cb39d34e61a4251b437f5138e2 : SUCCESS in 33m 04s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77ae3cd9bd8142eb8e8f75b31e45d1d0 : SUCCESS in 35m 03s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03414d96d2fd4346b4e3d3bacfc2e9b6 : SUCCESS in 43m 01s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/091c13c845bc40d4850f7366c00334c9 : SUCCESS in 38m 49s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/957324193b954d1494683860044b3a9a : SUCCESS in 40m 33s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/dd0bbed0f4ad419a9276aec3e6ddab51 : SUCCESS in 33m 04s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4447c5f8631346c0a656fe3ca5a80b33 : SUCCESS in 45m 12s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd2ec7429eaa486a953d6647d1f65214 : SUCCESS in 44m 20s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/4ebbb48648f147a2aef03984cc4eb325 : SUCCESS in 46m 44s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/c2cdf46165be46109d624e78896eefc3 : SUCCESS in 41m 30s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/e65dc297d2d844b28f0c7cca167a60ee : SUCCESS in 44m 50s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b3d4b72507845adb5e2494ee62b364e : SUCCESS in 40m 26s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/068bc5579b9148949897ad6efa93f701 : SUCCESS in 1h 26m 03s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/0ddd0241e8764544857711350aadf921 : SUCCESS in 1h 13m 42s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e231cc971384387b204ea7c8c6f2d7d : FAILURE in 1m 39s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/18ee571dff6b4e0b9d0a416f2ad11577 : FAILURE in 39m 40s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/794f4c9b6b674c8fb0bfc1ba25680320 : SUCCESS in 34m 35s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/2fd1b180a9b44c9a9bd44a058e6b2eb7 : SUCCESS in 40m 35s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b449de1f07e4a5cb15570cee3ad2e84 : SUCCESS in 32m 42s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/1e704d66bd5046e9bbb171f2f2636733 : FAILURE in 46m 14s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/865d39cc8f4247f8bec374d4c13ce9dc : SUCCESS in 37m 44s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/3a3a5b29c7cd450eb55523af5daef4a6 : SUCCESS in 32m 48s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/01ea25db1827469fa28beb5436e9c605 : SUCCESS in 39m 37s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/ddd5a946b7e041b9bf21478af719a408 : SUCCESS in 25m 35s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/1a105205878a402b83a627572879f49d : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1247e18e4912e63d7d74146fcb3d3f3e8d7c7997","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-02-03 14:44:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e39c354d3cd91b6ce1bdcf954fb11cd4ea29c90d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 14:44:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ddb145ce25d607e77b68faf42e396eef37eb2c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 16:34:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a1d4bfb0294e8a96e8459e6afa503b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65790d7d83ca425b88c9309ff489fade : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d03c1ebcb62c4b82ac96da860ee9ff28 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3569f92c97d4832a15f42a935df02c5 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/209401a96c9c4cadadddeb8753c197e7 : SUCCESS in 4m 56s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/44c2032534c9403882371cb91db8ab88 : SUCCESS in 6m 54s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/f3d5673fb4aa4aaeb828b63048b01015 : SUCCESS in 43m 36s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/615bcc3e014f446a8b466d431b3a6807 : SUCCESS in 24m 02s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f82e8d35cbc64c9a9eb2db25774cdf76 : SUCCESS in 38m 59s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/222d77b5920b401a92bce0aafdb81c23 : SUCCESS in 42m 17s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9f7aa05a1e7409d9b1348b1ac26ecfe : SUCCESS in 41m 36s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/2e6abcbf255141b5b64e62e8da2f5651 : SUCCESS in 44m 22s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/4eae9de3b6d5442982ca9793f32f4cc1 : SUCCESS in 42m 06s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/97d35ee63f3842318c2494e25d2c4d12 : SUCCESS in 34m 40s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eabeaa533ba4925acbadffac245b65b : SUCCESS in 42m 59s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/317bd5c5c2b341d3bda63d640e95af86 : SUCCESS in 41m 40s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/e2c70763a30a442c9540d8bd5b2c1063 : SUCCESS in 43m 01s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/862db4722da747238bbabb7d1e9b1d5d : SUCCESS in 34m 22s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/dd7e878402e94add859c2cde1c3c3424 : SUCCESS in 43m 30s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0344ada8b2d14cb2bc2cae9da720fbd0 : SUCCESS in 39m 16s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4fefdfc9dda46f5a98cd48c8668e2a6 : SUCCESS in 1h 21m 22s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/a0e8fa0fdbdc4408adf4af6d746cb4f8 : SUCCESS in 1h 39m 53s","accounts_in_message":[],"_revision_number":2},{"id":"bb2891fcfc7520c612ed4007bb8a9020363a5cbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 16:34:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0cbd4c73db1c7ca505f734b71795950cd5338404","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 16:35:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad2290917b92416ba60ccbe9e1a1094b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec592939e47a4d56ba994066bcdd1737 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e641f7e121f7044cef4fb7c80c776a73eee8f273","revisions":{"84b57f376b388d422156fd65b23267e5faf3ee0a":{"kind":"REWORK","_number":1,"created":"2025-01-31 10:19:59.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/05/940505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/05/940505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/940505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/940505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/940505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/05/940505/1"}}},"commit":{"parents":[{"commit":"1e4a29a5117029330b02627fdb9dc0f1e5255feb","subject":"Enable configuration of firmware_interfaces during install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/1e4a29a5117029330b02627fdb9dc0f1e5255feb"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2025-01-31 10:15:55.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2025-01-31 10:15:55.000000000","tz":600},"subject":"Update `uuid` to `id` in node_info to match module change","message":"Update `uuid` to `id` in node_info to match module change\n\nDue to change in ansible-collections-openstack baremetal_node_info\nmodule introduced by commit 902b2f8, node UUID is no longer stored in\n`uuid` field and `id` is used instead. Due to this  deploy-dynamic\nplaybook  may no longer work in certain scenarios. This change updates\nthe field to match the ansible module.\n\nChange-Id: I3e830e1c6caa85437c4aa6fbc95e5d384ee9562d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/84b57f376b388d422156fd65b23267e5faf3ee0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/84b57f376b388d422156fd65b23267e5faf3ee0a"}]},"branch":"refs/heads/master"},"e641f7e121f7044cef4fb7c80c776a73eee8f273":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-03 11:56:04.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/05/940505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/05/940505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/940505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/940505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/05/940505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/05/940505/2"}}},"commit":{"parents":[{"commit":"bf14155b8fa18d60e0865fb18aebd85a38fb2238","subject":"Fix ansible linters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/bf14155b8fa18d60e0865fb18aebd85a38fb2238"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2025-01-31 10:15:55.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2025-02-03 11:56:04.000000000","tz":0},"subject":"Update `uuid` to `id` in node_info to match module change","message":"Update `uuid` to `id` in node_info to match module change\n\nDue to change in ansible-collections-openstack baremetal_node_info\nmodule introduced by commit 902b2f8, node UUID is no longer stored in\n`uuid` field and `id` is used instead. Due to this  deploy-dynamic\nplaybook  may no longer work in certain scenarios. This change updates\nthe field to match the ansible module.\n\nChange-Id: I3e830e1c6caa85437c4aa6fbc95e5d384ee9562d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e641f7e121f7044cef4fb7c80c776a73eee8f273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e641f7e121f7044cef4fb7c80c776a73eee8f273"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
