)]}'
{"id":"openstack%2Fbifrost~940507","triplet_id":"openstack%2Fbifrost~master~Id1417f65e4e32183c21df62f65e73e4fc04afca7","project":"openstack/bifrost","branch":"master","topic":"doco-configdrive-fix","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2025-02-03 16:25:12.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-02-03 16:25:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1417f65e4e32183c21df62f65e73e4fc04afca7","subject":"Fix typo in CLI parameter spefifying config drive.","status":"MERGED","created":"2025-01-31 10:26:43.000000000","updated":"2025-02-03 16:26:51.000000000","submitted":"2025-02-03 16:25:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940507-doco-configdrive-fix","meta_rev_id":"f21b7413b4ea4320b57979936de80444e513402c","_number":940507,"virtual_id_number":940507,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-03 16:25:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2025-02-03 11:24:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-03 11:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-03 11:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-31 10:29:09.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-01-31 10:45:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-03 11:36:08.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"6b7e7f95867de283087c2b4adda1fdb19b15aeb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-01-31 10:26:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03bdd49959fe23f257263664a7ab84b0f9e179f3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-01-31 10:29:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"befcb16c99cbb0f1f10972d53578a35298df32d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 10:45:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8aa938e0e88f4b7c8ddb7b67388f844a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a168369a8054cf9bc5a10996e0a2390 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6fe9b03de95431ab8d697b6f1629f5c : SUCCESS in 6m 06s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/c86d1593004c4c57a217725e43050ed4 : FAILURE in 4m 42s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/ffc6a23bc39442aca94a67b707f6c4e4 : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2dc4d8821bcce309b4cf4c6ff1c1bfae62fef7fb","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-01-31 11:07:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"754ea7bac931f8f746405e8e1cfb3054057fa785","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-01-31 11:18:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d91b6c625abe3a2c542b19aad38be16b0f3f250e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 11:23:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/542806efb2294ea78967bacfc2125c88\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b26d31882a742538058521b1898503b : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f0a01bb5594eca993ed36c3fe3863a : SUCCESS in 3m 44s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/56004dbb24dc4b9d99f7c614208bb21c : FAILURE in 4m 14s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/3d9c635795f547a3a679627f998634dc : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edddc13f3e01346e7250999248df5aecfb4723fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-02-03 11:24:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2afe6ec117385800a5bbef03ca11f04e708cdcb1","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-02-03 11:25:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7dd00971548f4b7c54e20e5fc4d70b21b5a55622","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-02-03 11:36:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1208e3decf9b5b0e918a75d99bb75cfaa7d0d1fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 11:38:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0171cb4840048a2accac7c0d436431d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5daa984b0be44795a076f51ec6572cea : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c604a1ff1c864be8bc7e43ace2cd7093 : SUCCESS in 4m 59s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/83fcee53b0594635aa97adbdc85d85fd : SUCCESS in 4m 38s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/67696658982a4302abf75f5f2829b28f : SUCCESS in 2m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46a54972d6383e13f9f181af5ee6e7f714de46fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 14:43:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aba4647c56e32ba006887477ecafd6045f7fdfaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 16:25:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8ea55665987428b87fdffab982e2416\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a851344fcc046f68fb1954d06ec64ac : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83b05b158caa43388fe1de36968497dc : SUCCESS in 5m 13s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/ce06ba768e2645d79c3ab5376f506a71 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"7363fa6de9be3565c27b76317b61fbb214ab8d41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 16:25:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f21b7413b4ea4320b57979936de80444e513402c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 16:26:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fad7d5a16f04e2783ef37bf9dc2e9cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7559507960c34428aa49e99f1bd11f1e : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"196a5876d6b0d52e5d4e9f3b39c30946867bb40f","revisions":{"79844e1f67f94d672e7703f40fd810ea26d648c1":{"kind":"REWORK","_number":1,"created":"2025-01-31 10:26:43.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/07/940507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/07/940507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/07/940507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/07/940507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/07/940507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/07/940507/1"}}},"commit":{"parents":[{"commit":"a9dc56f4b925493a77555b1ef2426c5a502a7c9e","subject":"Fix leading and trailing whitespace in inspection_callback_url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a9dc56f4b925493a77555b1ef2426c5a502a7c9e"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2025-01-31 10:25:21.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2025-01-31 10:25:21.000000000","tz":600},"subject":"Fix typo in CLI parameter spefifying config drive.","message":"Fix typo in CLI parameter spefifying config drive.\n\nBifrost CLI takes `--configdrive` parameter while the documentation\nstates `--config-drive`. This commit fixes the typo.\n\nChange-Id: Id1417f65e4e32183c21df62f65e73e4fc04afca7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/79844e1f67f94d672e7703f40fd810ea26d648c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/79844e1f67f94d672e7703f40fd810ea26d648c1"}]},"branch":"refs/heads/master"},"196a5876d6b0d52e5d4e9f3b39c30946867bb40f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-03 11:24:50.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/07/940507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/07/940507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/07/940507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/07/940507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/07/940507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/07/940507/2"}}},"commit":{"parents":[{"commit":"bf14155b8fa18d60e0865fb18aebd85a38fb2238","subject":"Fix ansible linters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/bf14155b8fa18d60e0865fb18aebd85a38fb2238"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2025-01-31 10:25:21.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2025-02-03 11:24:50.000000000","tz":0},"subject":"Fix typo in CLI parameter spefifying config drive.","message":"Fix typo in CLI parameter spefifying config drive.\n\nBifrost CLI takes `--configdrive` parameter while the documentation\nstates `--config-drive`. This commit fixes the typo.\n\nChange-Id: Id1417f65e4e32183c21df62f65e73e4fc04afca7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/196a5876d6b0d52e5d4e9f3b39c30946867bb40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/196a5876d6b0d52e5d4e9f3b39c30946867bb40f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
