)]}'
{"id":"openstack%2Fbifrost~943180","triplet_id":"openstack%2Fbifrost~stable%2F2024.1~I4a6ba9bac2d134b22675fc56aa5dc4f51e71753c","project":"openstack/bifrost","branch":"stable/2024.1","topic":"backport-setuptools","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-03-06 14:34:59.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},"34076":{"account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"last_update":"2025-03-10 14:02:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a6ba9bac2d134b22675fc56aa5dc4f51e71753c","subject":"pip: Use SETUPTOOLS_USE_STDLIB if python \u003c 3.12","status":"MERGED","created":"2025-03-03 14:02:12.000000000","updated":"2025-03-10 14:02:56.000000000","submitted":"2025-03-10 14:02:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"943180-backport-setuptools","cherry_pick_of_change":924828,"cherry_pick_of_patch_set":2,"meta_rev_id":"5968db927047e2a4e5c4045d828ce709434df7bf","_number":943180,"virtual_id_number":943180,"owner":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-10 14:02:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2025-03-06 14:34:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-10 12:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-10 12:19:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-03 14:02:12.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-03-03 15:37:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-04 10:24:57.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-03-10 12:19:49.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"5c9b3669f80f39b6ebc3dbb46b477f3ab51b0ce7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2025-03-03 14:02:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ace3bc00c41d20df4fcacb15440b205cbd9b5826","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2025-03-03 14:03:06.000000000","message":"Topic set to backport-setuptools","accounts_in_message":[],"_revision_number":1},{"id":"6d66275562176abc025072faf4e075abe047c887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 15:37:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7449d5b022d4945982798281562814e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71a0d39a3154936a7a6a3189517e256 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0211b3275db34ef0bf25079c5662269f : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e721ffe4dfc4f41af425e2410c744ea : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d2caf549c44eed9893ba0f9f20a0af : SUCCESS in 5m 34s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/198acf15dc4f4beba8269269a208b6bc : SUCCESS in 7m 07s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/c8ac43a5e9594028a364b05b1ba7dd26 : SUCCESS in 44m 22s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38262ad461ad4cf0b8158e1ef8c1589c : SUCCESS in 55m 33s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/706d764771d64cbcaa6c21b8aa61a3d2 : SUCCESS in 43m 20s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/fdd520d17b2d4e96a8d3ce626a94b3f1 : SUCCESS in 52m 42s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/d348949900b948e1b9924116c4a4d59e : SUCCESS in 42m 34s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/5f2eab4dab394a799955efc9c9e3a2ef : SUCCESS in 36m 52s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a18e645ee4124ca1820328b605790811 : SUCCESS in 45m 51s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/87123b56e9e24c08ba730cbe51c47ef3 : SUCCESS in 25m 09s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/ddf16272fa6d437b9bc0b68f904f5b65 : SUCCESS in 36m 49s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/5d50a8a43f86455597af3fa1bf7f6d0d : SUCCESS in 41m 53s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc1123d76c9a4de7ae57ac14ed6a74a4 : SUCCESS in 37m 48s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31651e48b93249a8a5604967bc87ea87 : SUCCESS in 1h 19m 32s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/adf2aad423f24310a2030036e7505e29 : SUCCESS in 1h 32m 13s\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/74cb1ebb5fa74802be93cb1dd2407cb7 : SUCCESS in 44m 29s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bbad1660b9f4be2998b948e7d7dbfad : SUCCESS in 39m 01s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/07638ed69ed24bfb9bc5b6f9257f9902 : SUCCESS in 39m 07s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/513f33c07dc0454bbb6ca343b0e50df0 : SUCCESS in 37m 01s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/148ee7942cc446c49d437c3eac945951 : SUCCESS in 52m 46s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6993ed39489042ff8057a1764997fcf9 : SUCCESS in 36m 28s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/244399a743244d649a2bc73559a9593e : SUCCESS in 23m 39s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/d6ff6d0266a7406eaf13d892ab8847c8 : FAILURE in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec6ea9bd002286ec1e24a5540f7506d449f126e9","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-03-04 10:24:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"050513d17b222b5e16703786ab88dae6c0aae89f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-04 12:08:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efbfb7eb963ce0e4a2b73b82ac156de32f3b6cac","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-03-06 08:39:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9cb7ae150cf10291f07b87eb2e02a02c69975a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2025-03-06 14:34:30.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4b93c162cfdbfadcbfdb38a7df22168d31181947","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-03-06 14:34:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81f521ec07d87c74f9ffffd738f69d8ff4ae2f8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 16:12:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f9d356f478345d3b8e2313745be35b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7163415a2683428f80eaf14eabd8205c : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7a508ed6a1e404ba99681209e0ec549 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b18b4468370484c9a3fa017c41e90e7 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5719b827a9744a3ba1d834a0d21f9679 : SUCCESS in 6m 04s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/19f0ac9aa9054a94abf2a345a4b67f57 : SUCCESS in 8m 26s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/e857d2320de64d0b9a83d03e2ecab8cf : SUCCESS in 47m 36s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e3bd10c1c274ddfaa4007fee1b2fd1c : SUCCESS in 42m 09s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b378957de9a452bb74263a5c4e9b99e : SUCCESS in 42m 44s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/f224cc83ae2a4308b1c8d56e1784e664 : SUCCESS in 36m 06s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/e8d8be50314f4bdfba37ef4622a32dd4 : SUCCESS in 38m 49s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/c71880a8471c4bd48ae6fe834c8bec29 : SUCCESS in 42m 37s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/860c90e8c0c54d7fa9e3a87aeee2ad0f : SUCCESS in 55m 57s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/95e884216ef0491998a5bad56299adac : SUCCESS in 40m 35s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/60d0c4feaad64741a5b276878ebd0de0 : SUCCESS in 35m 32s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/70d02604a391439fab42a37a9104b2e0 : SUCCESS in 42m 07s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48a6bd03fc3d482480b8197787752f89 : SUCCESS in 35m 15s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24005492f5ff42b4ab0ee12357dbb8f0 : SUCCESS in 1h 08m 28s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/3cf1ba7865ac46ea8ce456cea890a182 : SUCCESS in 1h 32m 11s\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/489f2f413a7f43d9a9db88ff2598cf79 : SUCCESS in 45m 19s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9f851d636d1435fb5ec3355e22abbb8 : SUCCESS in 49m 14s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/0517f3dc56c2423fb587b6e38b741779 : SUCCESS in 45m 30s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69b74531a0c24275a5575a99f910fb69 : FAILURE in 44m 04s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/e3ef4ab38780429f9ade011a6a1695e7 : SUCCESS in 42m 35s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d4a650efcd0436a8aeded8dfdf8c1ee : SUCCESS in 46m 25s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/6fff02732e454ebb8c7dd7764435c2a5 : SUCCESS in 25m 14s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/2fa5d446116c4b008bd78a746b2a6f8d : FAILURE in 2m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6fcf76b5891eb2758740cd6fb2a67350ad8dcfa2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-03-10 12:19:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8ad987b8f85b20bb370842627a15a850fa8952d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 12:20:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00e7cf4bcd15753c73a363d986fd080bb9547fd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 14:02:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22762c2713084103b3ac50c87f89b428\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9291e205987d44468899b9f766390ef2 : SUCCESS in 1m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c901612ac33e455699ec78aafe13d663 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b7c6a705e864285b89625da5374bf29 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962ed11b66544c79b1d530f9ab081601 : SUCCESS in 4m 52s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/dd5854c7b25c44939b734aacee40b9f4 : SUCCESS in 6m 34s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/eb4eb81aa44d4b8599558c01be25566e : SUCCESS in 39m 49s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e47bb430934846c8995b8c8cc5a5ec8f : SUCCESS in 31m 49s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9480264774941f2afab084b29deb0c7 : SUCCESS in 43m 42s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/a5a13750e0164c3fbeacc36af42697d7 : SUCCESS in 21m 10s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/33562b802c654c5ab8d2ea691b0068a7 : SUCCESS in 19m 17s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/36fe9bcecee74a0ea4d1b8b05fda8105 : SUCCESS in 40m 36s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d48e86212924098a9636dd546b204e1 : SUCCESS in 33m 51s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/ad56aa21e6df457d919ebd34556676ca : SUCCESS in 40m 11s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/0d958f8847474ade81f78d9002ba0024 : SUCCESS in 33m 54s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/3fcf3a83c88d41088c592bb22a92905f : SUCCESS in 40m 21s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b4c9933c4c44c69abd2bdfd550f18e8 : SUCCESS in 37m 45s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/440908ab451543d6a90b71b6e085f394 : SUCCESS in 1h 16m 25s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/61766f4ee36f4b5dac84f82b6762d347 : SUCCESS in 1h 36m 44s","accounts_in_message":[],"_revision_number":2},{"id":"7c1ee7774e6fc756aef7905b0307a43df47ef6ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 14:02:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5968db927047e2a4e5c4045d828ce709434df7bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 14:02:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1def8fa0bd642dd9f5abe6c8085f0dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696e9fc30e8e4f8c98ca0e8244ecbe97 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02c65531332e046ae405a39d9c6350102edea8d2","revisions":{"2b69a311e43b1cfb7e431d8d3a161e312346881d":{"kind":"REWORK","_number":1,"created":"2025-03-03 14:02:12.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/80/943180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/80/943180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/80/943180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/80/943180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/80/943180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/80/943180/1"}}},"commit":{"parents":[{"commit":"c84cda2c2bad0ed05c3515306f540b0d99697d6f","subject":"Fix leading and trailing whitespace in inspection_callback_url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c84cda2c2bad0ed05c3515306f540b0d99697d6f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-24 08:54:45.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2025-03-03 14:02:12.000000000","tz":0},"subject":"pip: Use SETUPTOOLS_USE_STDLIB if python \u003c 3.12","message":"pip: Use SETUPTOOLS_USE_STDLIB if python \u003c 3.12\n\nCurrently using bifrost on Python 3.12 fails installation due to\nusage of that env variable.\n\nChange-Id: I4a6ba9bac2d134b22675fc56aa5dc4f51e71753c\n(cherry picked from commit 9825fecf384c6d5d4b09b94898ae746bef0b4784)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2b69a311e43b1cfb7e431d8d3a161e312346881d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/2b69a311e43b1cfb7e431d8d3a161e312346881d"}]},"branch":"refs/heads/stable/2024.1"},"02c65531332e046ae405a39d9c6350102edea8d2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-06 14:34:30.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/80/943180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/80/943180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/80/943180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/80/943180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/80/943180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/80/943180/2"}}},"commit":{"parents":[{"commit":"c84cda2c2bad0ed05c3515306f540b0d99697d6f","subject":"Fix leading and trailing whitespace in inspection_callback_url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/c84cda2c2bad0ed05c3515306f540b0d99697d6f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-24 08:54:45.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2025-03-06 14:34:28.000000000","tz":60},"subject":"pip: Use SETUPTOOLS_USE_STDLIB if python \u003c 3.12","message":"pip: Use SETUPTOOLS_USE_STDLIB if python \u003c 3.12\n\nCurrently using bifrost on Python 3.12 fails installation due to\nusage of that env variable. This backport is required to fix\nbifrost builds on Noble 2024.1 to allow to upgrade to Noble in\nthe Caracal-Epoxy upgrade cycle.\n\nNeeded-by: Ie12c43dc1de122c203edc28181366457ccd67834\nChange-Id: I4a6ba9bac2d134b22675fc56aa5dc4f51e71753c\n(cherry picked from commit 9825fecf384c6d5d4b09b94898ae746bef0b4784)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/02c65531332e046ae405a39d9c6350102edea8d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/02c65531332e046ae405a39d9c6350102edea8d2"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
