)]}'
{"id":"openstack%2Fbifrost~946116","triplet_id":"openstack%2Fbifrost~master~I64dddbaba37f3f92b999845575223f8df7b9b945","project":"openstack/bifrost","branch":"master","topic":"docs/update-howto","attention_set":{},"removed_from_attention_set":{"37841":{"account":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"last_update":"2025-08-22 18:03:47.000000000","reason":"Change was submitted"},"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2025-08-22 18:03:47.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-08-22 17:52:06.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},"35929":{"account":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"last_update":"2025-08-08 09:48:18.000000000","reason":"\u003cGERRIT_ACCOUNT_35929\u003e replied on the change","reason_account":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-08-08 14:39:40.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},"34205":{"account":{"_account_id":34205,"name":"Mahnoor Asghar","email":"masghar.bese15seecs@seecs.edu.pk","username":"mudo"},"last_update":"2025-08-22 18:03:47.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-08-22 18:03:47.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I64dddbaba37f3f92b999845575223f8df7b9b945","subject":"Clarify testenv and install usage order in testenv.rst","status":"MERGED","created":"2025-04-02 08:06:58.000000000","updated":"2025-08-22 18:04:54.000000000","submitted":"2025-08-22 18:03:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946116-docs/update-howto","meta_rev_id":"14fd6252b01caec913847a917a23f96b38badd88","_number":946116,"virtual_id_number":946116,"owner":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-22 18:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34205,"name":"Mahnoor Asghar","email":"masghar.bese15seecs@seecs.edu.pk","username":"mudo"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"recommended":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":1,"date":"2025-08-12 13:31:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":2,"date":"2025-08-22 17:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-08-08 09:48:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},{"value":2,"date":"2025-08-08 14:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34205,"name":"Mahnoor Asghar","email":"masghar.bese15seecs@seecs.edu.pk","username":"mudo"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":1,"date":"2025-08-22 17:52:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34205,"name":"Mahnoor Asghar","email":"masghar.bese15seecs@seecs.edu.pk","username":"mudo"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":34205,"name":"Mahnoor Asghar","email":"masghar.bese15seecs@seecs.edu.pk","username":"mudo"},{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-02 08:14:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-02 08:17:31.000000000","updated_by":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-04-02 08:17:31.000000000","updated_by":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"reviewer":{"_account_id":34205,"name":"Mahnoor Asghar","email":"masghar.bese15seecs@seecs.edu.pk","username":"mudo"},"state":"REVIEWER"},{"updated":"2025-04-02 08:19:35.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2025-04-03 09:35:03.000000000","updated_by":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"reviewer":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"state":"CC"},{"updated":"2025-04-07 10:12:07.000000000","updated_by":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"reviewer":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"state":"REVIEWER"},{"updated":"2025-08-06 22:55:36.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-08-12 13:31:25.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"}],"messages":[{"id":"e672b8c30aa88b5d8f8061d662dbeab148791917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-02 08:06:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"552907b82dec09bee9176640924206d94fd53402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 08:14:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/414599923ccf43f5bb8081ad46578d80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99a96642a0d84fefa2910108ec18ab8a : FAILURE in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db5707cbb8c476984d4db2f1b752609 : FAILURE in 5m 03s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/d8ff52c4c25f45ddaa7ccf5e287d10e1 : SUCCESS in 6m 14s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/a02a6e59557e401d94929ae956dde9f5 : SUCCESS in 2m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0eb064e1c9db2cb3ede2fdd61e2c054cb012351b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-04-02 08:19:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bbbd879e98a150ec6b003a1f073bada872fa9b83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-03 09:08:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cc018d8034c1557b1d951cc29e0f526aefa92863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 09:22:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7ae1075f0e94afb9cb21efa4a6575a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16cee1a594b2433e810c9b1473d4fd33 : FAILURE in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1444fdec75d94221a8880b415278af22 : SUCCESS in 7m 14s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/bb8c89838dca4784b2c443f33bed6a53 : SUCCESS in 6m 49s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/32408ff3c33e4be997e606b5e941b21c : SUCCESS in 3m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d99198ee35c55976ac76dc23f2ffd9aded72361a","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2025-04-03 09:35:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4ff67d4dd71b064310cc371a62f3e1a9b4f33e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-03 09:44:40.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5000d3f598655236814de541e473d74b212deafc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 09:56:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f6684a8a7734e44b9f28d3ac2659756\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04797ec5177e42adb5a4072dfeef5fe9 : FAILURE in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5152bcaa246d48aca49a8add398280af : SUCCESS in 6m 59s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/b40606c7d2464036906552e5f75d2b61 : SUCCESS in 7m 12s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/086d5754404944808571890d3ed5819e : SUCCESS in 2m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49e457ba71273fdbb11751473c39fcb11635b74a","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2025-04-03 10:46:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b46da5eb8d13c861858316a300cda22d941e0d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-03 12:18:56.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"efb779806024e3ecd68ce7d7f1e5426a5c445b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 12:29:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09d0b5df417d4adbbe3fee1197f41616\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/571c429304944972a699e2b4e6a9e830 : FAILURE in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec013f93c6e4edaa4eabef5d5b1b6c2 : SUCCESS in 8m 13s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/939a14d03a4b4f62bc34cb5f4a5200e0 : SUCCESS in 5m 33s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/3a1c28d8e074415eae0696a488e91992 : SUCCESS in 2m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fe13daa703f3e6839ec2e617b9535033d2c92405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-03 16:47:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ffe3ec16c3665127410b70d12ce2f0fa69d3aadb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 16:55:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8476353eb7924285af6bf353ac9d054c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bbe71f214974cae923c880d0e907fd1 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec9bdf690d04bfe8a593e68384f7782 : SUCCESS in 6m 36s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/3fd950ab9b7e4366a58d014286e3e5f8 : SUCCESS in 6m 25s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/fc94aa5114284a0d8407bd65abd6e833 : SUCCESS in 2m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5128f1442859cb90734ed7ed575351e9b8ecdc96","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-03 17:10:50.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5bb341f292e5871469d51173fe73d6caae2cfff1","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2025-04-04 09:00:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e127bb00bf33192b195a7676e426140c81754c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-05 18:03:39.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fe346587aa48d861f0e5906f6bad04c73bfdf6de","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-05 18:06:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5ac4a697ff951238b711bb12acfd88b5f8caa823","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-05 18:08:00.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"435c124d008e5c3c4d136f69e53ca07bf8c25873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 18:17:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe07599d680f44d8ac02b3fbdfe5b02e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93ee750702a84188ba16f4b22fae9edf : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820db9da4f9f4df2a791570bd704d360 : SUCCESS in 6m 21s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/c7f986a57a554b09b81caf14b1c075b4 : SUCCESS in 3m 12s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/d32ba9f235df463ab8fe2e7261eb1298 : SUCCESS in 3m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b632aa785cce2b46550516d521b864524aa369c8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-05 22:58:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"367dd7d487cfd3fb4ce914992e17087c7d61216f","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2025-04-07 10:12:07.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"17110cc819443e106f8ce3afca64c2e206e7c71d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-07 12:16:26.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6d4327cbc213e127fa0e68130ff3946de71bf332","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-07 19:45:27.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3678e136602f9758e649e6a9bb94596aa6762a34","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-07 19:54:44.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3a58c4aeb43cb82f1b200db88c0431d4524fc06a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 19:57:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff966c88a3b64bb8bd2860a4eccb9740\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f257a4fd6ff0422da0709273dcba4a1c : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d227645750c45d4be766e762b631b58 : SUCCESS in 10m 05s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/9e1afcda4de14452ad8deeda261e7b88 : SUCCESS in 6m 40s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/005a3b5ffec84967a9eb3a56748da0f4 : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5c8577b4eb63d473e1053b89669c24672bc783e8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-10 08:55:07.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"772af3ae24d70d7c18d83ddda0356d81790f61c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-10 11:33:10.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"2624daef697738f599ff273e7c1891097de16bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 11:40:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2482bf6ffef40569321c19d5806bac7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/632ff92205c843d69b656a8e4ae529c1 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72015e03a5a247739fc5550d67fdf4cb : SUCCESS in 6m 18s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/9c8d9d3943884111a5a03dfb590bce04 : SUCCESS in 5m 58s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/31dad70d7f8f4a6ea463a3eb418e08c7 : SUCCESS in 2m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ff574b351f8a20ec30f5d4ca89f80b7e10e463fe","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-10 11:40:17.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"bd41f427a464d0f55cf1246dc22f6bca7529b1af","author":{"_account_id":34205,"name":"Mahnoor Asghar","email":"masghar.bese15seecs@seecs.edu.pk","username":"mudo"},"date":"2025-04-17 11:04:09.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"5030fc4f2a21e01aec69f369d8716ef06a457f6a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-04-17 14:20:55.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"6116d35eb7620181baa0135191e819e491189396","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-04-22 10:57:28.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a16e76845efc40ca7babcfb61ba0f0cc43e275fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-22 14:24:59.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"1667d91330cc48e3589d33f4de97bc6e8d9a93ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 14:32:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86770375d1c94f7f85165495320469bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fdfb4216bde4ca08307149328c7d0b7 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d47133882cc4a688ca0e3d5f88475f5 : SUCCESS in 4m 22s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/3a366c01285342f19890a336b092d335 : SUCCESS in 3m 55s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/68abeee203774729aa0acec96f2eec4e : SUCCESS in 3m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"394ebc4d513a031bdd3d30b502c39d2d02625dfc","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-04-22 16:08:43.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"52014817450a96a113857c6438e0cc684b28b13f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-04 13:15:06.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":9},{"id":"dbc6b39900096056929c16b89c7882ff76609c23","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-08-06 22:55:36.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"82ea03d4bab4f07177cd57bdd44e075e8b1d8bfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-08-07 21:19:44.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"1684c56ad59e336baaeab8259849db19a0c4ea73","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-08-07 21:21:03.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"16c61731e01c7d5fa4c09dbe10a92fe6a20606f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 21:31:48.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18af1aa6e3fa414190c8f8f88d972e31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877c5e98312b41bb992debfb78e946b3 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/618239f9b2d845869954c61559f73e25 : SUCCESS in 3m 27s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/1ad6e868ae1542609c657cf1394196e9 : SUCCESS in 7m 38s\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/856ec66e188f41228b296b86b4e63bbc : SUCCESS in 3m 09s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":10},{"id":"3ee7fce5a0d150ad457c0dc8dba8060e52983064","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2025-08-08 09:48:18.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"88e7be889ba6c2880b7df2b3bd22aeb0c44c5bf9","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-08 14:39:40.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"cfe2f359c91b8e057706e71f446226c976c3348f","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-08-12 13:31:25.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"19dbdee1345c32077b3b0da788e205551fdf8f23","author":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"date":"2025-08-12 20:28:05.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1e6c4c914cfb151ce4dcde0251ee7d439debd61d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-08-22 17:52:06.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"5f30fdd2636883bb88754059ac9f369c0b95e45a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 17:52:34.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"0fa3e57893e304f946176d8997cb47fe30a4e1d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:03:47.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/829f0c8089d9414a8c9e1cd0824c6a29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86aedfe298f94f1d98dbce20bba47310 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff852ec86634e20831cdd8098ade83b : SUCCESS in 9m 56s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/134bca60150a46b8bf4022b78268686c : SUCCESS in 8m 32s\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":10},{"id":"7b1c2198c2ddde4398199e438269527fc0148d01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:03:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"14fd6252b01caec913847a917a23f96b38badd88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:04:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23890eb7275d4732b7c5c8f2a180b905\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adee590e99c5473b8c6f10169c151f17 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"86366ec4543b6c886020abbb062dcb4f4dd91124","revisions":{"cf9f839a522e4d90074c8359d1925a6f833362f9":{"kind":"REWORK","_number":1,"created":"2025-04-02 08:06:58.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/1"}}},"commit":{"parents":[{"commit":"7798b495cbf1152f49f62a108d17fcb254dc34ac","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7798b495cbf1152f49f62a108d17fcb254dc34ac"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"subject":" Improve howto.rst with updated instructions","message":" Improve howto.rst with updated instructions\n\nThe how-to guide lacked clarity on managing and providing nodes\nin a Bifrost test environment. Users may struggle with the process\nif they are unfamiliar with Ironic’s workflow.\n\nThis update enhances the documentation by:\n- Clarifying node management and provisioning steps.\n- Emphasizing the need to check node power states before management.\n- Providing additional details on verifying Ironic installation.\n- Adding troubleshooting section for node enrollment and provisioning.\n\nThese improvements ensure a smoother experience for first-time users\ndeploying nodes in a test environment.\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cf9f839a522e4d90074c8359d1925a6f833362f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/cf9f839a522e4d90074c8359d1925a6f833362f9"}]},"branch":"refs/heads/master"},"e06dba3b96af9a3fbb4baf101ea034054797e503":{"kind":"REWORK","_number":2,"created":"2025-04-03 09:08:39.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/2"}}},"commit":{"parents":[{"commit":"7798b495cbf1152f49f62a108d17fcb254dc34ac","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7798b495cbf1152f49f62a108d17fcb254dc34ac"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-03 08:50:13.000000000","tz":-420},"subject":"Refactor documentation: Move updated instructions to testenv.rst and troubleshooting to troubleshooting.rst","message":"Refactor documentation: Move updated instructions to testenv.rst and troubleshooting\nto troubleshooting.rst\n\nThe initial commit for the updated instructions and\ntroubleshooting details were done in the howto.rst. Based on reviewer feedback,\nthis commit correctly organizes the documentation.\n\nChanges include:\n- Moving updated deployment instructions to testenv.rst.\n- Moving troubleshooting details to troubleshoot.rst.\n- Ensuring howto.rst remains focused on its intended purpose.\n\nThis improves the structure and clarity of Bifrost documentation,\nmaking it easier for users to follow.\n\n(previous commit message: \"Improve howto.rst with updated instructions\")\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e06dba3b96af9a3fbb4baf101ea034054797e503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e06dba3b96af9a3fbb4baf101ea034054797e503"}]},"branch":"refs/heads/master"},"46fa167afafd2a3778d2b2ce9ff508fd69f1efa6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-04-03 09:44:40.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/3"}}},"commit":{"parents":[{"commit":"7798b495cbf1152f49f62a108d17fcb254dc34ac","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7798b495cbf1152f49f62a108d17fcb254dc34ac"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-03 09:44:40.000000000","tz":0},"subject":"Refactor documentation: Move updated instructions to testenv.rst and troubleshooting to troubleshooting.rst","message":"Refactor documentation: Move updated instructions to testenv.rst and troubleshooting to troubleshooting.rst\n\nThis commit reorganizes the Bifrost documentation for better structure and clarity.\n\nChanges include:\n- Moving updated deployment instructions to testenv.rst.\n- Moving troubleshooting details to troubleshooting.rst.\n- Ensuring howto.rst remains focused on its intended purpose.\n\nThis improves the structure and clarity of Bifrost documentation,\nmaking it easier for users to follow.\n\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/46fa167afafd2a3778d2b2ce9ff508fd69f1efa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/46fa167afafd2a3778d2b2ce9ff508fd69f1efa6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6c32fa0e3b1d3950d7ee01170ec18edecf9b667c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-04-03 12:18:56.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/4"}}},"commit":{"parents":[{"commit":"7798b495cbf1152f49f62a108d17fcb254dc34ac","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7798b495cbf1152f49f62a108d17fcb254dc34ac"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-03 12:18:56.000000000","tz":0},"subject":"Move updated instructions to testenv.rst and troubleshooting to troubleshooting.rst","message":"Move updated instructions to testenv.rst and troubleshooting to troubleshooting.rst\n\nThis commit reorganizes the Bifrost documentation for better structure and clarity.\n\nChanges include:\n- Moving updated deployment instructions to testenv.rst.\n- Moving troubleshooting details to troubleshooting.rst.\n- Ensuring howto.rst remains focused on its intended purpose.\n\nThis improves the structure and clarity of Bifrost documentation,\nmaking it easier for users to follow.\n\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6c32fa0e3b1d3950d7ee01170ec18edecf9b667c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/6c32fa0e3b1d3950d7ee01170ec18edecf9b667c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"de1823ebef932241d41d72a7acdb153c000d997e":{"kind":"REWORK","_number":5,"created":"2025-04-03 16:47:58.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/5"}}},"commit":{"parents":[{"commit":"7798b495cbf1152f49f62a108d17fcb254dc34ac","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7798b495cbf1152f49f62a108d17fcb254dc34ac"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-03 16:44:13.000000000","tz":-420},"subject":"Move updated instructions to testenv.rst and troubleshooting to troubleshooting.rst","message":"Move updated instructions to testenv.rst and troubleshooting\nto troubleshooting.rst\n\nThe initial commit for the updated instructions and\ntroubleshooting details were done in the howto.rst. Based on reviewer feedback,\nthis commit correctly organizes the documentation.\n\nChanges include:\n- Moving updated deployment instructions to testenv.rst.\n- Moving troubleshooting details to troubleshoot.rst.\n- Ensuring howto.rst remains focused on its intended purpose.\n\nThis improves the structure and clarity of Bifrost documentation,\nmaking it easier for users to follow.\n\n(previous commit message: \"Improve howto.rst with updated instructions\")\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/de1823ebef932241d41d72a7acdb153c000d997e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/de1823ebef932241d41d72a7acdb153c000d997e"}]},"branch":"refs/heads/master"},"1bcaf3f03b8320c4bcf7ea24c8461728a30e8d53":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-04-05 18:03:39.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/6"}}},"commit":{"parents":[{"commit":"7798b495cbf1152f49f62a108d17fcb254dc34ac","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7798b495cbf1152f49f62a108d17fcb254dc34ac"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-05 18:03:39.000000000","tz":0},"subject":"Moved updated instructions from howto.rst to testenv.rst and troubleshooting  to troubleshooting.rst","message":"Moved updated instructions from howto.rst to testenv.rst and troubleshooting  to troubleshooting.rst\n\nThe initial commit for the updated instructions and\ntroubleshooting details were done in the howto.rst. Based on reviewer feedback,\nthis commit correctly organizes the documentation.\n\nChanges include:\n- Moving updated deployment instructions to testenv.rst.\n- Moving troubleshooting details to troubleshoot.rst.\n- Ensuring howto.rst remains focused on its intended purpose.\n\nThis improves the structure and clarity of Bifrost documentation,\nmaking it easier for users to follow.\n\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/1bcaf3f03b8320c4bcf7ea24c8461728a30e8d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/1bcaf3f03b8320c4bcf7ea24c8461728a30e8d53"}]},"branch":"refs/heads/master","description":"Edit commit message"},"94f2dc3ce690f24d551f10d8048493f3b31721f2":{"kind":"REWORK","_number":7,"created":"2025-04-07 19:45:27.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/7"}}},"commit":{"parents":[{"commit":"7798b495cbf1152f49f62a108d17fcb254dc34ac","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7798b495cbf1152f49f62a108d17fcb254dc34ac"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-07 19:43:22.000000000","tz":-420},"subject":"Clarify usage of master branch and SSH key types in testenv.rst","message":"Clarify usage of master branch and SSH key types in testenv.rst\n\n- Added note clarifying that the example uses the development branch (master)\n  and how to switch to a stable branch.\n- Clarified that RSA is used as an example for SSH keys, but other key types\n  like ed25519 are also supported.\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/94f2dc3ce690f24d551f10d8048493f3b31721f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/94f2dc3ce690f24d551f10d8048493f3b31721f2"}]},"branch":"refs/heads/master"},"11482fd44a57056c3326860a1ab53549a8981f75":{"kind":"REWORK","_number":8,"created":"2025-04-10 11:33:10.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/8"}}},"commit":{"parents":[{"commit":"7798b495cbf1152f49f62a108d17fcb254dc34ac","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7798b495cbf1152f49f62a108d17fcb254dc34ac"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-10 11:13:40.000000000","tz":-420},"subject":"Clarify usage of master branch and SSH key types in testenv.rst","message":"Clarify usage of master branch and SSH key types in testenv.rst\n\n- Added note clarifying that the example uses the development branch (master)\n  and how to switch to a stable branch.\n- Clarified that RSA is used as an example for SSH keys, but other key types\n  like ed25519 are also supported.\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/11482fd44a57056c3326860a1ab53549a8981f75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/11482fd44a57056c3326860a1ab53549a8981f75"}]},"branch":"refs/heads/master"},"36a009ba8af1bbbbe79db08d043f9767e3e10177":{"kind":"REWORK","_number":9,"created":"2025-04-22 14:24:59.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/9"}}},"commit":{"parents":[{"commit":"7798b495cbf1152f49f62a108d17fcb254dc34ac","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/7798b495cbf1152f49f62a108d17fcb254dc34ac"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-22 14:22:22.000000000","tz":-420},"subject":"Clarify testenv and install usage order in testenv.rst","message":"Clarify testenv and install usage order in testenv.rst\n\nReordered the documentation to keep all `testenv` usage and arguments together.\nMoved the `install --testenv` section below to maintain a logical flow:\n- First set up the test environment\n- Then install Bifrost services into it\n\nThis improves clarity and aligns with reviewer feedback.\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/36a009ba8af1bbbbe79db08d043f9767e3e10177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/36a009ba8af1bbbbe79db08d043f9767e3e10177"}]},"branch":"refs/heads/master"},"86366ec4543b6c886020abbb062dcb4f4dd91124":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":10,"created":"2025-08-07 21:19:44.000000000","uploader":{"_account_id":37841,"name":"Queensly Kyerewaa Acheampongmaa","display_name":"Queensly Kyerewaa Acheampongmaa","email":"qacheampong@gmail.com","username":"queensly"},"ref":"refs/changes/16/946116/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/16/946116/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/16/946116/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/16/946116/10"}}},"commit":{"parents":[{"commit":"a43b5bd0f49a86c71894d8bb6f62b32e4e1c02ef","subject":"Move all CS9 jobs to non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a43b5bd0f49a86c71894d8bb6f62b32e4e1c02ef"}]}],"author":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-04-02 07:53:12.000000000","tz":-420},"committer":{"name":"Queensly Acheampongmaa","email":"qacheampong@gmail.com","date":"2025-08-07 21:19:40.000000000","tz":0},"subject":"Clarify testenv and install usage order in testenv.rst","message":"Clarify testenv and install usage order in testenv.rst\n\nReordered the documentation to keep all `testenv` usage and arguments together.\nMoved the `install --testenv` section below to maintain a logical flow:\n- First set up the test environment\n- Then install Bifrost services into it\n\nThis improves clarity and aligns with reviewer feedback.\n\nChange-Id: I64dddbaba37f3f92b999845575223f8df7b9b945\nSigned-off-by: Queensly Kyerewaa Acheampongmaa \u003cqacheampong@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/86366ec4543b6c886020abbb062dcb4f4dd91124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/86366ec4543b6c886020abbb062dcb4f4dd91124"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
