)]}'
{"id":"openstack%2Fbifrost~955181","triplet_id":"openstack%2Fbifrost~master~I872aa87ff030b0dcf0f7b4031fd228819ad1ee62","project":"openstack/bifrost","branch":"master","topic":"update-uc-pinned","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-08-06 23:33:47.000000000","reason":"Change was submitted"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-08-06 23:33:47.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-08-06 23:33:47.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-08-06 16:49:24.000000000","reason":"\u003cGERRIT_ACCOUNT_10239\u003e replied on the change","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"hashtags":["ironic-week-prio"],"change_id":"I872aa87ff030b0dcf0f7b4031fd228819ad1ee62","subject":"Updated pinned upper-constraints for Python 3.9","status":"MERGED","created":"2025-07-16 11:17:31.000000000","updated":"2025-08-06 23:34:57.000000000","submitted":"2025-08-06 23:33:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955181-update-uc-pinned","meta_rev_id":"668d2798dca43b6840705677c25337b525e89da8","_number":955181,"virtual_id_number":955181,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 23:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-08-06 22:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-06 16:49:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2025-08-06 22:50:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-16 12:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-21 15:06:40.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-07-21 15:08:51.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-07-21 15:50:24.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2025-08-06 22:50:59.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"3cc833f77cc372cb4ab0539b18e9c3c111285a7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-16 11:17:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ce50384a7041ab6f8e9ab5f33a0f03c170dc8fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 12:29:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb5d81f1db7a4c348d22bf9d80980170\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26bbe3c80c364575bc65602eba53a482 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26716ffae4d9465e976abb365dbf13ae : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8859f275017042488735b12b6b8f0cfb : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09a29d697f5342ffacf769d7354a09d5 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcd29bdbaa642c594c850d945d78643 : SUCCESS in 6m 52s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/0531b2d303c84243910756f9090dc6d7 : SUCCESS in 7m 38s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/98a2941965d14dfb8b131b356acf69b0 : SUCCESS in 45m 54s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e416d00807cf4c3f97264c4694ebab07 : SUCCESS in 44m 30s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37e774ac8582467d86051491484dd154 : SUCCESS in 39m 06s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2881e02ff1a84f00816ebbca6cc4a8d2 : SUCCESS in 44m 28s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8526808e8124f4ab3fecf3aa1dd241d : SUCCESS in 40m 25s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/fb1135ab37d64e7d88ea3cdf69398d09 : SUCCESS in 46m 59s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/e8dc69aecfce4c1f92e5d28bc9b6fcf6 : SUCCESS in 39m 44s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/1ea2cbdc2d31488f836bae255ea87dc1 : SUCCESS in 40m 19s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73de03c891474b8fbdf4dd7c3105d16d : SUCCESS in 38m 27s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e90d4b8aa234772ae0fe43a3b94c6db : SUCCESS in 41m 52s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/2e3262d6dd80464cb76922fd261188ac : SUCCESS in 37m 14s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/5972b502e77e4c498bcad0354391228f : SUCCESS in 36m 07s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/6c091d5c8e974bc9b4daec434951713f : SUCCESS in 42m 41s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4ac1d28f3b54889b5f1d685fd5bff88 : SUCCESS in 40m 58s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d28daf4e4a84fd8abc310660a1b8692 : SUCCESS in 1h 09m 27s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/dffb822978dd4285bab510f9f088a4aa : SUCCESS in 1h 07m 20s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cf8805cc35045c198d3608bcad1cdc5 : SUCCESS in 37m 55s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/3b5c76e9f1204120b170bbb27b1b161b : FAILURE in 38m 59s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a5bbc83ed224816aa7162ae70d3dfa3 : SUCCESS in 31m 16s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/15938f67eb8746dba3cce2e6e305159a : SUCCESS in 51m 21s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0c3feac8b7c494c872deb841cf0f290 : SUCCESS in 31m 16s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/f7e76d7326ac4536bdd918231ddada02 : SUCCESS in 34m 46s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bf43223574a496084f12c206b1d99b8 : SUCCESS in 38m 44s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/73011ea881634711b526b5bb3d8321b6 : SUCCESS in 28m 04s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/5f1a0ef922df43db8fee7406b460c47c : SUCCESS in 20m 21s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/9a930a26b4784328bb70ca1b0f46ad09 : SUCCESS in 27m 27s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/76c6c173289c4ba791d2303248d59071 : SUCCESS in 1m 41s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"42d5ce19d0b78e85f4b425a416be7eaaa7f39d74","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-16 13:04:03.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"7abae7b29c199f44da2f3da426314be2da99f15c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-07-21 15:06:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ffe618f6c54d188b008bef325b6f06263a982841","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-07-21 15:08:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"60018fe5b0cc7db4e499c06eea3b2b13b1fca611","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-07-21 15:50:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f10354018b2038383ac4ce2db20813bebc67ee54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 15:50:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c688061ca2c5a554b26dd7dd986e620900e329e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 17:16:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02281e612b8d4f638bfdfd57e3753be6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2e4bab156b4639980f1b4fbacfe205 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/222296b305a24b49b6aab39e0db66bda : SUCCESS in 2m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b883adddb227404a919b7c12d80bde67 : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f30bb5da7b0c418d9b2c42ab67c735cb : SUCCESS in 7m 08s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/18e8bed6e76442a6b2db02e39426aa41 : SUCCESS in 4m 06s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/474d6b668f9f40c6853c856bd8799738 : SUCCESS in 52m 20s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c824a4b455214292b356b4d35f936b04 : SUCCESS in 47m 28s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab0ca8bd50584e5aaf3c1fa652707bbe : SUCCESS in 27m 53s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3004b7042d82465986a232aea79634ad : SUCCESS in 20m 57s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7b796d342484cc98ee3703a5e2fbdcd : SUCCESS in 30m 22s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/6ba59593472f420c8c412fdc86ed3b07 : FAILURE in 11m 49s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/a48043cb76e64d99968b3aa38ca5aba8 : SUCCESS in 40m 50s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/9b0faada3699479d8209120b2657375c : SUCCESS in 35m 20s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4da7a2c52a1e4243b02a5e38b2d4c68c : SUCCESS in 30m 00s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25a656d68ced413ca77704d91420cac9 : SUCCESS in 35m 40s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/f9abf8fde8344b4ab9cafe8a655bf1a6 : SUCCESS in 19m 08s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/7ea21207571b40aca350da176671267a : SUCCESS in 17m 26s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/76ac088ea26b48eeae036aefec4e9777 : SUCCESS in 44m 06s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ba11b4407e74c9ca5cb50b0473b75d0 : SUCCESS in 38m 00s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dafbb7c3dd174e959d89f0512eb0d2dc : SUCCESS in 1h 20m 36s\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/43a7d61586d14a8daee78cd6fdae2761 : FAILURE in 27m 42s\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"df3d5776cf85f6a0b83fd4567278c4adff3c5762","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-22 08:19:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c73ada37679c29ec0264873dc221f2a20ae69c5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-22 08:26:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1cf203d145f72290ff9bf3709043644aecf3ce52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-22 08:27:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"55d0daf792846596ab3aa2162a29644d17a66d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 08:27:23.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/bifrost on branch master.  The\nproblem was:\n\n  The nodeset \"ubuntu-jammy - openstack/ironic-prometheus-exporter\" was\n  not found.\n\nThe problem appears in the the \"bifrost-integration-tinyipa-keystone-ubuntu-jammy\" job stanza:\n\n  job:\n      name: bifrost-integration-tinyipa-keystone-ubuntu-jammy\n      parent: bifrost-integration-tinyipa\n      nodeset: ubuntu-jammy\n        - openstack/ironic-prometheus-exporter\n      vars:\n        enable_keystone: true\n        enable_tls: true\n        test_driver: redfish\n  ...\n\n  in \"openstack/bifrost/zuul.d/bifrost-jobs.yaml@master\", line 89","accounts_in_message":[],"_revision_number":2},{"id":"2d7904d28dc2e800d770074f90e1a3a77798d27d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-22 08:34:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d8c21006360cea5fa69dfac7d74e0c6b3290460c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 09:52:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/996298a104eb46d59c8896d1d880bd60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3378c9f0221f4f0586f1adc8db14dc6c : SUCCESS in 1m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea0eebfdc70d4774adbebca78d69c7ab : SUCCESS in 2m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d4cb4c3bd4f4de9b0fd5eec1edfb4fc : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/134b388b856e43de8c59f84bb3f66cf4 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c5c8a2d373408d905dc870a3fb69be : SUCCESS in 6m 57s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/75ad1c7b22ae4893b0009acd9bf1bf4b : SUCCESS in 7m 02s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/afdfcec28b524c228ad8f96be531156f : SUCCESS in 44m 19s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1659908f11240408c6ce02ac94afdbb : SUCCESS in 21m 19s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2b63fcc79ab42369500d1fa21a9eee7 : SUCCESS in 33m 10s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b38637d7ef19469583c0bcaeb84d49bd : SUCCESS in 42m 45s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d19679bde33d4438a50c82555b571ff8 : SUCCESS in 34m 04s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/6a01f5cdac7f4284ad949d622e560701 : SUCCESS in 40m 58s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/d3533e485fe54278956cfb1a39a36c39 : SUCCESS in 38m 14s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdeea365b11e4a5d924280c0ff516698 : SUCCESS in 18m 50s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cd0e72d6ebe4db2b4fb4d87f4b521d7 : SUCCESS in 19m 18s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/ca92e2c831d4486a89f2e571885a810b : SUCCESS in 47m 26s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/7db4de24079f4f9c991bdb8c0a32814e : SUCCESS in 42m 17s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/d16dca36f3c04d569be2befc69b7dec2 : SUCCESS in 38m 30s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/689cfe15ca7549cead375628562724db : SUCCESS in 40m 32s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac4f2cc6c68549259338a07252866496 : SUCCESS in 1h 10m 03s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/7524122f568b43b1ba7c9f27768b42ff : FAILURE in 8m 57s (non-voting)\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/233a8ce8c79e4d66b2817d55ae5bdabd : FAILURE in 7m 54s (non-voting)\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c48f1c45bef8469ab6e3824b81e0b0c9 : SUCCESS in 43m 53s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/e446db6012084b29b9b7980561cb06d0 : FAILURE in 46m 45s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1690c0d8b57b4ae69bbe961ac1e26ea5 : SUCCESS in 38m 47s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/f60eade7d6014c3caea025d88aedbd23 : SUCCESS in 46m 21s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b837b5e42a2f45d98046210312bfcceb : FAILURE in 29m 52s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/3a6606bf6ad94e658e5d34c3055946e0 : SUCCESS in 43m 51s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7483ef14e6b4972b01d73fc90e422f1 : SUCCESS in 36m 21s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/65cd6eb3f16044338a7a315814367c7e : SUCCESS in 20m 48s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/9a5f02ba6dc04217b9f46ff861ca5c92 : SUCCESS in 38m 30s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/02bfd4a6ef98454ea47099a80be97a52 : SUCCESS in 28m 20s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/1591d4760574420ea640f2c5d912e613 : SUCCESS in 2m 39s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":3},{"id":"3e2572f1506227d01357e105e33ef5e13c41c776","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-08-06 11:16:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb8782784e91ef294074ce7fa4c39253374c373f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 12:01:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"54e819f1c31e130ed12b6278d6853c835db25699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 12:41:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"846c1f43882c3fbd0c5662081693a680d29554e2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 12:41:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7e46de491cdcbaccb9f735b53ec176be96834b43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 14:13:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/983c4aa1c6894205a7c50e6a01276472\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/676a6c12203c4d81a55cae7006c24c4d : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c539c4a18c504240aec51037fb92fc86 : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23889b85941841018c91d131fb8a3aef : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea1555b87a8145a9b6df9dc982db99c6 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc450258cee4c3a9e1efc4079d2f153 : SUCCESS in 8m 35s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/9186a2a38ff845d3a5f416d02c9ca4a8 : SUCCESS in 8m 02s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/5f0d329e3a4f42618958bf3cd8473cba : SUCCESS in 24m 32s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b707525100894f0db5b10b67ac9b8c0e : SUCCESS in 38m 43s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02b75daccddc4fb48813e589c58fb4a8 : SUCCESS in 20m 30s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49fe10da81ff460992c3ad5a5d7bc428 : SUCCESS in 44m 25s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f688c00af2b40069d78f0d833a391dd : SUCCESS in 39m 05s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/ebdcc1fc93f0491b921c762a1432af94 : SUCCESS in 22m 59s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/09fc9c24b78f4bac8732ecc62a767163 : SUCCESS in 39m 50s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9297423d994848859c8141122f75181e : SUCCESS in 46m 48s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68664e95fc604cb2a151a117183ccf1a : SUCCESS in 49m 27s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/4a2fb084b8d44dcd911198eaa7ed7251 : SUCCESS in 40m 30s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/93f79e5d362541c988ccad63edc8eff5 : SUCCESS in 17m 34s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/602485994cf94a959752a099dfb205b1 : SUCCESS in 44m 50s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3e0a98070fc49d5a8cd2df789f4a8af : SUCCESS in 40m 56s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5519865676b6469fbf54314a5f39f305 : SUCCESS in 1h 21m 19s\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/1d8e548fe58c465881d0a5ef04958255 : FAILURE in 10m 42s (non-voting)\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/e9fd975acc0043fc8b5fed7862e88e9a : FAILURE in 52m 52s (non-voting)\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2af43d9e172144599f3af9774454708d : SUCCESS in 1h 23m 14s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/1ecef94eb45d4932936638bb2374552b : FAILURE in 48m 27s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7b42a7bf73c4be4a9e9dd67d0448c59 : SUCCESS in 41m 05s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/16de2477d80d4755b5e788f069f58424 : SUCCESS in 40m 14s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ef130bdf61b43dd873f1e29e1f265c5 : SUCCESS in 38m 33s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/bb8fe40bf6c34f7b969baf526b59ce25 : SUCCESS in 44m 41s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1844987e36384d9d834080d33e336927 : SUCCESS in 43m 47s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/2481c4c86f7d4087846cc3fc004aa24e : SUCCESS in 41m 47s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/e92cd666179f4c65a5d5880ecfa1c30f : SUCCESS in 17m 29s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/848888af0a6948d1ba9d5fbce41f68ad : FAILURE in 22m 29s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/a5f4467c4b6c4965b90fe560743f8875 : SUCCESS in 3m 32s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":4},{"id":"4b639254ed4626309fa11e37171a21b69af46a1b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-08-06 16:49:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"64be9f519a09ba7c292025f02c962a23725a30a9","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-08-06 22:50:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"69d2287ee1edabbbe51630c1894e890cd65bbde4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 22:51:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"842eadb9ef8355f8698f0464b9e8fdb24f307b6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 23:33:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31cde8c9b87a4e838d6c3631752fef2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a60b340867ad4c729212f5ea3ec97cb9 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf29f9e8f7554f72831b560af8b5b022 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3123212fddfb4d17aee054cc94638f53 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c878aff43b714d079773fa5624c59131 : SUCCESS in 6m 47s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/166124ef450146a0b71eeceff0a91145 : SUCCESS in 6m 36s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/90d4ba41d4ab461291c6fde2d7914e49 : SUCCESS in 23m 50s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e4728299bbe4cd9b2dc55bf69278aad : SUCCESS in 33m 17s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/794b27a54bb148dfb25d78fe5bd6c9b5 : SUCCESS in 36m 44s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f35c416f961a4fe79b7f130746841398 : SUCCESS in 35m 10s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56802938a62d42c78394525218d1bdaf : SUCCESS in 18m 58s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/c76a9983aaf74dd5a70f7141be078a71 : SUCCESS in 31m 22s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/b2d23f77e7974baab50a069405fa898a : SUCCESS in 20m 58s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b1530250c694a71b03271563beb3558 : SUCCESS in 38m 13s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25a461858f404b22ad693a5743f59cea : SUCCESS in 34m 46s\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/51b25c3bb40d4c3195069fec3042c94d : SUCCESS in 19m 00s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/a1b4c2e4f40e45fc9e8d72b105c8085b : SUCCESS in 20m 08s\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/18cc12ae73004b3693a514ef8498d3c1 : SUCCESS in 34m 03s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9d72e0279324a36ac6034365fb05d9d : SUCCESS in 32m 17s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86358a9521a9438592a63c7b46218b1d : SUCCESS in 41m 32s\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":4},{"id":"37def53b8cec898500a279e33d562306d3a2036d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 23:33:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"668d2798dca43b6840705677c25337b525e89da8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 23:34:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44580b44329d4d4485a513d9cb3f7178\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5087e0f953544d9da21f3f5de173fb81 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e0c98b6a72a2da1331b222e7804532f0c9f84758","revisions":{"db0c751f6e5407cbef20b3951866705c3db96c75":{"kind":"REWORK","_number":1,"created":"2025-07-16 11:17:31.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/81/955181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/81/955181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/81/955181/1"}}},"commit":{"parents":[{"commit":"034609e075106b42ed26ee184c72d5d6731d5eef","subject":"Release note for ironic-inspector deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/034609e075106b42ed26ee184c72d5d6731d5eef"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-16 11:16:52.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-16 11:17:27.000000000","tz":120},"subject":"Updated pinned upper-constraints for Python 3.9","message":"Updated pinned upper-constraints for Python 3.9\n\nPreparing to moving cs9 based jobs to non voting\n\nChange-Id: I872aa87ff030b0dcf0f7b4031fd228819ad1ee62\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/db0c751f6e5407cbef20b3951866705c3db96c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/db0c751f6e5407cbef20b3951866705c3db96c75"}]},"branch":"refs/heads/master"},"0c2b4a517c68eab91f5b98fe9283a6e887c3f0c3":{"kind":"REWORK","_number":2,"created":"2025-07-22 08:27:10.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/81/955181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/81/955181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/81/955181/2"}}},"commit":{"parents":[{"commit":"034609e075106b42ed26ee184c72d5d6731d5eef","subject":"Release note for ironic-inspector deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/034609e075106b42ed26ee184c72d5d6731d5eef"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-16 11:16:52.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-22 08:27:05.000000000","tz":120},"subject":"Updated pinned upper-constraints for Python 3.9","message":"Updated pinned upper-constraints for Python 3.9\n\nPreparing to moving cs9 based jobs to non voting\n\nChange-Id: I872aa87ff030b0dcf0f7b4031fd228819ad1ee62\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0c2b4a517c68eab91f5b98fe9283a6e887c3f0c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/0c2b4a517c68eab91f5b98fe9283a6e887c3f0c3"}]},"branch":"refs/heads/master"},"afe8663563ff4c963c4360291c4d92058fd1e637":{"kind":"REWORK","_number":3,"created":"2025-07-22 08:34:59.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/81/955181/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/81/955181/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/81/955181/3"}}},"commit":{"parents":[{"commit":"034609e075106b42ed26ee184c72d5d6731d5eef","subject":"Release note for ironic-inspector deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/034609e075106b42ed26ee184c72d5d6731d5eef"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-16 11:16:52.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-22 08:34:36.000000000","tz":120},"subject":"Updated pinned upper-constraints for Python 3.9","message":"Updated pinned upper-constraints for Python 3.9\n\nPreparing to moving cs9 based jobs to non voting\n\nAlso moving keystone cs9 jobs to non-voting as keystone\ndropped support for Python 3.9\n\nChange-Id: I872aa87ff030b0dcf0f7b4031fd228819ad1ee62\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/afe8663563ff4c963c4360291c4d92058fd1e637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/afe8663563ff4c963c4360291c4d92058fd1e637"}]},"branch":"refs/heads/master"},"e0c98b6a72a2da1331b222e7804532f0c9f84758":{"kind":"REWORK","_number":4,"created":"2025-08-06 12:41:28.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/81/955181/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/81/955181/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/81/955181/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/81/955181/4"}}},"commit":{"parents":[{"commit":"034609e075106b42ed26ee184c72d5d6731d5eef","subject":"Release note for ironic-inspector deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/034609e075106b42ed26ee184c72d5d6731d5eef"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-16 11:16:52.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-08-06 12:41:22.000000000","tz":120},"subject":"Updated pinned upper-constraints for Python 3.9","message":"Updated pinned upper-constraints for Python 3.9\n\nPreparing to moving cs9 based jobs to non voting\n\nAlso moving keystone cs9 jobs to non-voting as keystone\ndropped support for Python 3.9\n\nChange-Id: I872aa87ff030b0dcf0f7b4031fd228819ad1ee62\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e0c98b6a72a2da1331b222e7804532f0c9f84758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e0c98b6a72a2da1331b222e7804532f0c9f84758"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
