)]}'
{"id":"openstack%2Fbifrost~955660","triplet_id":"openstack%2Fbifrost~master~Ibaf56d9460ef34922fe32e63680c58c42c59ac4b","project":"openstack/bifrost","branch":"master","topic":"update-uc-pinned","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-08-07 00:16:49.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-08-06 16:50:07.000000000","reason":"\u003cGERRIT_ACCOUNT_10239\u003e replied on the change","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"hashtags":["ironic-week-prio"],"change_id":"Ibaf56d9460ef34922fe32e63680c58c42c59ac4b","subject":"Move all CS9 jobs to non-voting","status":"MERGED","created":"2025-07-23 08:18:29.000000000","updated":"2025-08-07 00:18:38.000000000","submitted":"2025-08-07 00:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955660-update-uc-pinned","meta_rev_id":"8f2b4c8f78eebcc3c428b5df7499e04c6a4cdedc","_number":955660,"virtual_id_number":955660,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-07 00:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2025-08-06 22:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-06 16:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2025-08-06 22:50:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-23 09:21:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-06 11:17:10.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2025-08-06 22:50:38.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"c5ed0aa530a0bcda8cfc26ed47e9914e7a77fa7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-23 08:18:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0528a169a3a73df6b42e60b4446578f80ed7efa1","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-23 08:18:46.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"b9468a8bc5e247fd518a9f5217b1f7f5f65f83fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 09:21:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cb00c2273144df3aa03dfd00dd034a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be1573a770748bd8f7e7e93153114d4 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f9ad7d923e649f2bd397f9b0d046316 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/623915304d8e45079a412b58407868f5 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad5195e19e5441e1a4e1d5ad029e4227 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae344b92d6947c89286b5f0a6b26a2d : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c991a158e74c426da111801b62257e11 : SUCCESS in 3m 02s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/f9bfd2e14cde4a7cab8596809cc290a9 : SUCCESS in 5m 29s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a0e2d4d5d69437aaf22fc7e989d64a1 : SUCCESS in 40m 50s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e507408eb67f42af93387cd5bf5a2b23 : SUCCESS in 26m 58s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23811357080b401c9cea130587ef22d2 : SUCCESS in 39m 17s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e722148e6a8b413c9230252490198c47 : SUCCESS in 28m 50s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/87f3aeb21f0d43e7905a6a5b786a8050 : SUCCESS in 37m 47s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/51080b9ac766464db992e53bd1c3ebe0 : SUCCESS in 38m 43s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30942260e1fc4377b25b796be268f344 : SUCCESS in 28m 42s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97d7d92fcd5246acb2e32d6cf43326c2 : SUCCESS in 36m 58s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/79e050c081c744d6b9b6dd72326fef12 : SUCCESS in 28m 14s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a86bed62a4be45bdb886587b3d1e29a5 : SUCCESS in 28m 13s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7be16eb73b394e14804659db00fcdae7 : SUCCESS in 39m 36s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/fbd3b379da554e2699d860fb2268b5f7 : SUCCESS in 32m 13s (non-voting)\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/968d5c8a3659414c8d8886e678c56c3e : SUCCESS in 31m 09s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/a3178a9a5afe4842a4c5247c9e9becbb : SUCCESS in 30m 06s (non-voting)\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/75b72684931b4b1c889b4ad587049c53 : FAILURE in 7m 56s (non-voting)\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/d67ae6cb456f404b8bfd6f6de37a03c3 : FAILURE in 6m 53s (non-voting)\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c5ff690f3ce4d2c86d36be66cfb09d7 : SUCCESS in 53m 31s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/4df32b4e1bde400dac9ba0b031bbb04d : FAILURE in 41m 30s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19b4baf61f7046ad89e5a94da04b9e9b : SUCCESS in 30m 27s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/a971612fc2fd4bb5b326f62a34cf0429 : SUCCESS in 33m 48s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/826da033ca2d4ead863c3c951ba5d6dd : SUCCESS in 37m 23s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/3352490d633b4147b0d68be4da0e5557 : SUCCESS in 43m 38s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55e9920b99d2443aa6e3f5e143cea4c3 : SUCCESS in 22m 39s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/da71bc957c1142698c2fac7ee6b16d19 : SUCCESS in 19m 12s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/937ccd6bb2834f79873b766334661164 : SUCCESS in 39m 23s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/cc2ae090d0354fc69d2cd4940b27e9c1 : SUCCESS in 13m 07s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/a17162b9046546c6a6f779254a1eb400 : SUCCESS in 2m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"61afccf7a3aa4240a390df655c4195b3b746a4d6","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-08-06 11:17:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"916286cc77e002d5cc4d102da8efebeb743743db","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 12:01:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b364f1762beaa6bc0df9d02ae38c4024fc1554e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 13:11:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"448bc90b43fa6451a269cd3438e14d5a4c14474e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 13:12:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f2e8adbd4c22eb8fa1837cdd1b4ab17939eea5ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 13:13:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9ab6461958a5acd33938e30b9ee90d6f1ed5e7c3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 13:13:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"78f3931a7b318dc2ea63c91c0ec4089fa977f953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 14:42:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ca51158d4ca42ef9cab1dadb85c03bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c48d0fe3d83245e7b258568aeff65b39 : SUCCESS in 1m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/814e468f50b84920a0518eed929e8de8 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5a51e0c302c463b989f7fdd963bf613 : SUCCESS in 1m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/280f1e4c8bc241ccbfdfcb11ca84ac98 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab29bc4651bb4f609984bc72be430af0 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58c6a7639ca74dad933674412782fbe1 : SUCCESS in 1m 41s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/8e686ec7fc1b4106ae8aa50b559789d3 : SUCCESS in 3m 32s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d5c314b450e4e8e9732cc328727a743 : SUCCESS in 37m 16s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a301d21ffc04118ba484481907562b9 : SUCCESS in 20m 59s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/689dd28d2ebd4c0c9a4961bff395b4dc : SUCCESS in 43m 25s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a1848bcfb334353a24a8a2891ad26da : TIMED_OUT in 1h 00m 54s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e0a3d6f2c62480ba285fb24c3b755cd : SUCCESS in 33m 42s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/3598ddc9a5d24645ac9299d40a52e4f9 : SUCCESS in 21m 05s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db63520b094e40348e85b1721c41e9aa : SUCCESS in 36m 20s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7eeabf7cb1234b00859f95e2e9595eb3 : SUCCESS in 35m 32s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/c3e1eeaf297b4759806ad2252e755d7a : SUCCESS in 35m 04s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d0513981ccb495e92f1d617a929b97e : SUCCESS in 35m 11s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79740a34cdea462186874b0225ef08e5 : SUCCESS in 1h 23m 32s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/d307db87b39741288f314800a381f46d : SUCCESS in 37m 45s (non-voting)\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/70686228d3054dc9af18b3a379792d40 : SUCCESS in 41m 44s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/ab6f6e1f1c7e4a5bb2c36e8d3a1315b2 : SUCCESS in 19m 17s (non-voting)\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/843cb04692454430b1871ea27d15ba7b : FAILURE in 9m 49s (non-voting)\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/6c55465149d9479ca9cabd6c11a5809d : FAILURE in 24m 04s (non-voting)\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23088aba7acf4215b22f896e6598323e : SUCCESS in 1h 11m 35s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/249029aae51b464cac13e7b2274d7ff4 : FAILURE in 42m 53s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88515dd368344006b0291dc5225cc855 : SUCCESS in 33m 08s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/10167cd625044adfaf2b3703addeece3 : SUCCESS in 27m 33s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6b0315f2cfc4a38949fe4ebfe3fdb59 : SUCCESS in 36m 32s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/824d1bd993d046aea0a224a63ebd3756 : SUCCESS in 38m 27s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ecf451a52104237a374bbe934afcc8f : SUCCESS in 39m 58s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/b85a1af1f9ee4b4091cf8d0b560625d2 : SUCCESS in 23m 08s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/f41525359d074d20b0f82061b84d6c46 : SUCCESS in 19m 04s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/54b2a8a81c1247bc8d31fa28f4e2b108 : FAILURE in 18m 46s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/e9d3ccd9b80c4b9cb2df18dae124ad7a : SUCCESS in 3m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":4},{"id":"236b4c65b97304863e94b0f2655b7190f4152f8a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 14:44:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4dbf6c85b4f18db73a77d49e172315de00034ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 16:11:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0676f475714442ea6e0bf7dadc8baf2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/974384caf4d94c04948c34df176f891c : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbc7d1895663453d84f2ff20ee34cae1 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bf91402cc284b1d828e7e23763f195c : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd5d5b49454d4edfa8cf7bc914e297a6 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbfd7132d1544a9198052b3de3f2d0b8 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e93ee3cffd194e07b22f4e509fa8c7ef : SUCCESS in 1m 38s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/8baab9e08ec843a18a5c3a508657bccb : SUCCESS in 9m 25s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3453c644fe8c44258ea4c3944a65f8e5 : SUCCESS in 40m 45s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7088caf7df70402696259fd12f5ee6ae : SUCCESS in 38m 25s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3785579b43b14e38b5dc9761020ef376 : SUCCESS in 41m 40s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fcc4a61c7314bc090bbc197c82e989a : SUCCESS in 40m 24s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/d35cf548999944248e748fd8c3d36b91 : SUCCESS in 32m 40s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/2da31082a3a84f6fa30f7af6b0ed452f : SUCCESS in 45m 06s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74ff5354bca74a0abd44c7c2192b56d5 : SUCCESS in 40m 28s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd3f190693f5439d80bb17d87a4075bd : SUCCESS in 19m 54s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/8927539f3d134d80834663cc395b9cbc : SUCCESS in 33m 31s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03cf334f898d4bbc91feaa255e627b31 : SUCCESS in 37m 52s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2a58e392d494aab8be37262c9fd83eb : SUCCESS in 1h 19m 32s\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/1cf9a578e4d748188f3e0cb364fe575f : SUCCESS in 43m 34s (non-voting)\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/085e91cc61c147f298b8693a6880b2e9 : SUCCESS in 41m 19s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/ef3e6be994364586bd62ce76d00ab86b : SUCCESS in 47m 54s (non-voting)\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/7c6c0f03e6064c1c990918eef577d88b : FAILURE in 5m 33s (non-voting)\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/26568064166f43a3ae148eb46412ed1e : FAILURE in 53m 33s (non-voting)\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f4675ab7ae84c1fa8d2209e5eabfec4 : SUCCESS in 40m 23s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/60778a521fa447429fb0c7e78e395b4e : FAILURE in 44m 44s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6842d16404ab4d62930dfc7431256c01 : SUCCESS in 25m 06s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/3bc511be6eaa4d6daba730e5dd1a5304 : SUCCESS in 27m 18s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b44f67c9ca24d069961582b0c005a7f : SUCCESS in 47m 31s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/a56fa625cb2b469cbade44526f709d60 : SUCCESS in 45m 20s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53d7b4719d7d43da855ae2ccd0387806 : SUCCESS in 40m 42s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/05f4e9095eb54bf2996cfbc5e18d18d7 : SUCCESS in 37m 51s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/b417fe1581cb4a9bb3d9f456b5b87625 : SUCCESS in 39m 08s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/11130daeae8d401c9142ce2ea9ed4352 : FAILURE in 21m 46s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/81b2acf653484606a7e05f47bc17d475 : SUCCESS in 3m 31s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":4},{"id":"65fab103cc41c1cc6f2d454d5d768b2c947afafa","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-08-06 16:50:07.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f102e3927c7d56c933c3f90b4e1ea42a261edbb6","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-08-06 22:50:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"63e697ddd2f330810ebc20725a850c0433f07500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 22:51:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"568e12a3684332fa59d520930b5f08bfe9afc112","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 00:16:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b4d5bfbfc13404c822c308daf90e985\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4658ad17be1d4d43aded70e5f3c402ef : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a234135cdc4a4b7bad040466cb1b74d9 : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c12817786b743aba0bb6d4e042de702 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03db5d29c602495a96e02a418e655603 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03795e5157814019bd143a9d25e31377 : SUCCESS in 4m 02s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/5493ef5afb064de1b06d4601c3a459a3 : SUCCESS in 8m 56s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa88ede1f0274e66a10591b7047e0d97 : SUCCESS in 19m 35s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eca77bd1af1a46959182d0b3aca922da : SUCCESS in 34m 50s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9db70b82404d4762a6d18295c0099f91 : SUCCESS in 45m 33s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49b99ed0dcb64965bb4d50dc4567eb1c : SUCCESS in 36m 23s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/b2d9a01914054743a70a273436ad96d6 : SUCCESS in 34m 43s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/a2e793a182b34e2db8e8d3ea6063b6e7 : SUCCESS in 23m 07s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7239c93ec3254f52baf68d816563c107 : SUCCESS in 31m 02s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07058f8d84c645b4a1700962e76b523b : SUCCESS in 42m 33s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/a07a08d69e144a0981b630a19ed11a21 : SUCCESS in 37m 52s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d71d4202a2d843afb12e0f39259994a4 : SUCCESS in 36m 23s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb42c0960cf7454795b55503b3cef709 : SUCCESS in 1h 20m 07s\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":4},{"id":"fb2ea489ac5e0a750a05479af5d58b162f4645f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 00:16:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8f2b4c8f78eebcc3c428b5df7499e04c6a4cdedc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 00:18:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed86599bc8404cc29d0683a6dbc0b74a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4015ce92b4d47929232c1a66be38073 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82a946afe4be470cb39094a3f5910267 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a43b5bd0f49a86c71894d8bb6f62b32e4e1c02ef","revisions":{"a1d5b14cc65592aae9d027897f9791c19d29b471":{"kind":"REWORK","_number":1,"created":"2025-07-23 08:18:29.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/955660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/60/955660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/60/955660/1"}}},"commit":{"parents":[{"commit":"afe8663563ff4c963c4360291c4d92058fd1e637","subject":"Updated pinned upper-constraints for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/afe8663563ff4c963c4360291c4d92058fd1e637"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-23 08:15:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-23 08:18:24.000000000","tz":120},"subject":"Move all CS9 jobs to non-voting","message":"Move all CS9 jobs to non-voting\n\nWith the upcoming support removal for Python 3.9 we can\u0027t\nguarantee that CentOS Stream 9 jobs will keep working.\nMoving all of them to non-voting and deprecating support.\n\nChange-Id: Ibaf56d9460ef34922fe32e63680c58c42c59ac4b\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a1d5b14cc65592aae9d027897f9791c19d29b471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a1d5b14cc65592aae9d027897f9791c19d29b471"}]},"branch":"refs/heads/master"},"b2f0741d8334c029f306c290238509fa5099ce2e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-06 13:11:33.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/955660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/60/955660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/60/955660/2"}}},"commit":{"parents":[{"commit":"e0c98b6a72a2da1331b222e7804532f0c9f84758","subject":"Updated pinned upper-constraints for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e0c98b6a72a2da1331b222e7804532f0c9f84758"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-23 08:15:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-08-06 13:11:33.000000000","tz":0},"subject":"Move all CS9 jobs to non-voting","message":"Move all CS9 jobs to non-voting\n\nWith the upcoming support removal for Python 3.9 we can\u0027t\nguarantee that CentOS Stream 9 jobs will keep working.\nMoving all of them to non-voting and deprecating support.\n\nChange-Id: Ibaf56d9460ef34922fe32e63680c58c42c59ac4b\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b2f0741d8334c029f306c290238509fa5099ce2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/b2f0741d8334c029f306c290238509fa5099ce2e"}]},"branch":"refs/heads/master","description":"Rebase"},"e66b51b401bba67e3462cb69ef63de1f7714ee9d":{"kind":"REWORK","_number":3,"created":"2025-08-06 13:12:27.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/955660/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/60/955660/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/60/955660/3"}}},"commit":{"parents":[{"commit":"afe8663563ff4c963c4360291c4d92058fd1e637","subject":"Updated pinned upper-constraints for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/afe8663563ff4c963c4360291c4d92058fd1e637"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-23 08:15:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-08-06 13:12:17.000000000","tz":120},"subject":"Move all CS9 jobs to non-voting","message":"Move all CS9 jobs to non-voting\n\nWith the upcoming support removal for Python 3.9 we can\u0027t\nguarantee that CentOS Stream 9 jobs will keep working.\nMoving all of them to non-voting and deprecating support.\n\nChange-Id: Ibaf56d9460ef34922fe32e63680c58c42c59ac4b\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e66b51b401bba67e3462cb69ef63de1f7714ee9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e66b51b401bba67e3462cb69ef63de1f7714ee9d"}]},"branch":"refs/heads/master"},"a43b5bd0f49a86c71894d8bb6f62b32e4e1c02ef":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-08-06 13:13:10.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/955660/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/60/955660/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/60/955660/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/60/955660/4"}}},"commit":{"parents":[{"commit":"e0c98b6a72a2da1331b222e7804532f0c9f84758","subject":"Updated pinned upper-constraints for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/e0c98b6a72a2da1331b222e7804532f0c9f84758"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-23 08:15:44.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-08-06 13:13:10.000000000","tz":0},"subject":"Move all CS9 jobs to non-voting","message":"Move all CS9 jobs to non-voting\n\nWith the upcoming support removal for Python 3.9 we can\u0027t\nguarantee that CentOS Stream 9 jobs will keep working.\nMoving all of them to non-voting and deprecating support.\n\nChange-Id: Ibaf56d9460ef34922fe32e63680c58c42c59ac4b\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a43b5bd0f49a86c71894d8bb6f62b32e4e1c02ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a43b5bd0f49a86c71894d8bb6f62b32e4e1c02ef"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
