)]}'
{"id":"openstack%2Fbifrost~957847","triplet_id":"openstack%2Fbifrost~master~If3ff89ec4128fe2c12fa9c470da4f04163f5f667","project":"openstack/bifrost","branch":"master","topic":"deprecate-bullseye-support","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-29 11:46:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-09-02 20:58:04.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-09-01 10:08:19.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2025-09-02 20:58:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3ff89ec4128fe2c12fa9c470da4f04163f5f667","subject":"Recover the CI","status":"MERGED","created":"2025-08-19 08:38:51.000000000","updated":"2025-09-02 21:00:16.000000000","submitted":"2025-09-02 20:58:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957847-deprecate-bullseye-support","meta_rev_id":"a7475e64ec3ae8f225df48d6f2075d5eb36a87a8","_number":957847,"virtual_id_number":957847,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-02 20:58:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2025-09-02 19:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-08-29 11:46:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-01 10:08:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2025-08-29 11:48:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-08-29 13:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2025-09-02 19:22:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 09:20:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-19 11:20:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-20 09:07:40.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-08-29 11:22:11.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2025-08-29 11:48:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-08-29 13:33:32.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2025-09-02 19:22:30.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"6c779bbd2ab1b0e8ff0752968456e02f5afc6d38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-19 08:38:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99343a8e9d73f90da8dc0a80afd6d3e269f0bd12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 09:20:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b806b097466a47f7b63bbc367bccf128\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07d992210a884c799448b535a8f44257 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6213fc465aaa43f794a711929b4f1ccd : SUCCESS in 2m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/974ae0d31d8943caaf86416e48dcbdd4 : SUCCESS in 2m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd21b3c07b1f487fba14f4ef5614641e : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f67dbf81d774a75b95f8b27cb3690bf : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2160be0b7b948cf83287aca2e57fe8e : SUCCESS in 2m 00s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/98ec3987a2494881a652e0b810536eea : SUCCESS in 4m 49s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0d26287a5ec48e0950ec935a3431573 : SUCCESS in 20m 32s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/890beed4a6bc427983c6f40150971a0a : SUCCESS in 29m 18s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f61721e31e794783a9b97e9b8251ee34 : SUCCESS in 22m 18s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6be66534e5e74fc29a554d64b40c1052 : FAILURE in 6m 00s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/732782931e2142f0b44ba32d54655f6a : SUCCESS in 19m 24s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/204bb627767d44e38705359ba5e7cb46 : SUCCESS in 22m 25s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/118d3ce20dc64f2f9101516b1325f422 : FAILURE in 5m 06s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/cf2af5f5cc764012a985e9f3335a3a0d : SUCCESS in 18m 21s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b98cb367f0d1484baa529eafadd01551 : SUCCESS in 37m 06s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6531c26b020484f80a8d9918479b476 : FAILURE in 21m 24s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/21044abd055b44cca45565c771cb9c01 : FAILURE in 7m 30s (non-voting)\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/5f7a1fb6fe7f4b1bb4d6e76948cba67b : FAILURE in 12m 37s (non-voting)\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/11c9119e01c54915abba14fca3a29138 : FAILURE in 6m 05s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/80cfdb88efe240e190485f790bc71fd9 : FAILURE in 13m 05s (non-voting)\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/e780e80c38cc48a58bea65d3fb596824 : FAILURE in 5m 12s (non-voting)\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/0f801fca2ddc42d0900171c8c085754e : FAILURE in 26m 11s (non-voting)\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8246a7589f4e43e3ba55d13642df4880 : FAILURE in 23m 22s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/c35b92ae3a1442d68f8e85f6f17aa807 : FAILURE in 13m 02s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13a192f2677d45479dfbba0d7c53bef1 : SUCCESS in 39m 50s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/ecd1c16ddbfe48579c20e3b2d64cef9c : FAILURE in 13m 31s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efef2f77d229481892943be2c94209b7 : FAILURE in 12m 25s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/515b1a062feb42a3aa5a1cc2d3359cf3 : FAILURE in 6m 17s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/634e5cf228af431f903eaa301861c3fc : FAILURE in 9m 11s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/348fb1512de34e3d98753fb7e2e6a373 : SUCCESS in 19m 56s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/72baa5762e7a49b881b405ef944475e9 : FAILURE in 6m 27s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/52abd8b291624fe989409dac1159da60 : SUCCESS in 20m 42s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/f52b0e2e15ed435a97861edad387596f : SUCCESS in 2m 03s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"e3bbf61527d3e8b5a50c47adb9c05d13eea152d2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-19 09:29:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef6ce26bd53abb6b86298a6dd23f97f0ca73f2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 10:28:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8807f0aa66e4738bfc5457dad01f13f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/996160ace2af4ee88b435691ed1dd950 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/419c66affc7f418ca1b709c50d434470 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cabee4644c5142cfb6ebb78e935c82e2 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3619e97550334e88993fd621cb9c515f : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f81e66ff0de42288862f439106462bd : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0d7823291ae4812becaa405185e1771 : SUCCESS in 3m 12s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/a875a5c12c1d4979a155853646cff3e4 : SUCCESS in 7m 08s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98e9087d7e184a16a98200cfc151ae58 : SUCCESS in 23m 19s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be198ef09e1346879833a2eaf2d242c2 : SUCCESS in 20m 02s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a60c2c82bf842a8a3e35000305137c4 : SUCCESS in 36m 21s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/403f2c73b7a14e3daea89298cf0e282f : FAILURE in 10m 32s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/74cb05101a194f9fb878394d61299e85 : SUCCESS in 22m 02s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/348404ac1310462e9259b52f9fa459e7 : SUCCESS in 36m 32s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d0742476cfc4b12aa93e91ab0e67ea3 : FAILURE in 11m 29s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/09dbc912e6024a12ba1904f5894c1ce8 : SUCCESS in 21m 56s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76118517cc784905bc3865be500a20b2 : SUCCESS in 20m 59s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f76c9604dd847f185ce008664dc0176 : FAILURE in 42m 57s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/b831313681384d1598f706506f115cdf : FAILURE in 4m 34s (non-voting)\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/5ff42ee6b5eb4b0384f0d3274d83834d : FAILURE in 12m 22s (non-voting)\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/df0f30a64e684f27a107c9f6808eb64b : FAILURE in 6m 53s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/d771831fd3734b46b3fdc3d4ecb849e4 : FAILURE in 11m 33s (non-voting)\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/2e7d45fbf43046ebb3cfc3c5db1bcaae : FAILURE in 9m 21s (non-voting)\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/d18ee88e5c9841ecbb56d057dbc1ddad : FAILURE in 51m 47s (non-voting)\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/128a99a70a1c470eaf3023d41e6284c9 : FAILURE in 41m 18s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/e8af5f758d32478180c8d2e2be25a2c7 : FAILURE in 11m 48s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75b25753dc874e1ea82a65d88534a92f : SUCCESS in 36m 15s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/a077f24d116f441ca96d63e64bbee20e : FAILURE in 12m 43s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5a9a6463b8e4ee18d5cd35601076ca0 : FAILURE in 8m 27s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/c0b8d1028a9c4dcb9d47f44b8c59072a : FAILURE in 12m 35s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69248a6c57734be1b5e1ee7867761135 : FAILURE in 11m 19s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/6dd14cbd987e470f9b9b097d339eb6ea : SUCCESS in 37m 59s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/2626d84bde904953bd7784e7c9a3d3ad : FAILURE in 10m 08s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/178bc4d7c7e6487fa82a335436bc4ee9 : SUCCESS in 31m 05s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/493884cc75c640ef969f38f059357854 : SUCCESS in 3m 39s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"d2a60dec8c4fe4a112019409e2b8f01fc4c18fc4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-19 11:20:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2d8d9feb0781aef89c7d4e1fb75ed8e4112ac9a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-20 09:07:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9040f2b08b5198f3c04dc41b6cf77d5ddaade63f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-20 13:10:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0a604d2ed51801bbc6d4c5267b07674c63d9437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:57:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72e7893b49f84285b33c174c81f50122\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb9a8b7629884fa18f0796ea1438c283 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0626c79e7c8644478a4e5cb9f8901f99 : SUCCESS in 2m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a9d797e5c244a5fafb232e778b303bd : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ca2883e3920414e952e58dd86bbba82 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae70a6cdd6e140a5a5ec56eac7eed3f2 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad13a0c0e9324a0eaec4d4a03aa6c64d : SUCCESS in 2m 28s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/6ad6165b22384a16814ad7ce76641f37 : SUCCESS in 6m 46s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d396c3bec424460b3b8a899d9ad3161 : SUCCESS in 20m 42s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c96a2f536e14a6ba48bb771f58a3d5f : SUCCESS in 37m 51s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43e04dffed0445f399cb0fe9bdfe794d : SUCCESS in 20m 02s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/638937f79cb94321a408a1799ffe350a : FAILURE in 6m 17s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/1b8a869d6a9b4b63b7797573d05485f2 : SUCCESS in 38m 59s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0a6fe12ae2942d498a156779bc445a7 : SUCCESS in 40m 16s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51aeaf87e1714b7cbe66b3f08ca4e809 : FAILURE in 8m 13s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/e400b7fb52d5444a808b9416b24067ab : SUCCESS in 19m 10s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e65882f4b6a4734a8416ffdcf443066 : SUCCESS in 37m 30s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3751dc1532e54518ae04c6c6f54ed9e9 : FAILURE in 38m 46s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/0120a4ca72b3499e9e8c34edcb7e9877 : FAILURE in 7m 39s (non-voting)\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/7fc809f12fb441e58d73a5abe422d4b4 : FAILURE in 13m 04s (non-voting)\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/bacc6de9b2ff43f996e1a5d8f3c06d80 : FAILURE in 12m 29s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/9ec8f9b7234f47bc8fe9f4a424239dec : FAILURE in 6m 01s (non-voting)\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/678e8ecc9f4e46a5bcc0ea94983d9af7 : FAILURE in 12m 01s (non-voting)\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/e5bb0eb191d447c49678678a25d27fce : FAILURE in 24m 54s (non-voting)\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1f18ad48be3479588df12b5f53868cd : FAILURE in 23m 39s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/f72a48ec7f704385ae949e6b3e5bd2e9 : FAILURE in 10m 23s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91a896536caf46438d13893c2977fc45 : SUCCESS in 37m 07s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/0bf779ee5fbd471091eee9a86b7778c0 : FAILURE in 12m 07s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ac7d6041bee4ed886285b162923aa32 : FAILURE in 11m 58s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/db3854d1208846e8ae68aa4263fc209c : FAILURE in 12m 25s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fc5dfe00b0b419faa17456d6b145e0d : FAILURE in 6m 32s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/63d61a9c82654a579078497989199099 : SUCCESS in 41m 30s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/c4c630ce66674a9cba646249968a5246 : FAILURE in 6m 15s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/e2dfed0c37994127b87f22af69e1400b : SUCCESS in 13m 45s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/6b99b148b1f44e9987dc118af86d7690 : SUCCESS in 4m 15s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"6cbaef2e3938e06a2863920f50c0cdfb42cbdaff","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-28 07:21:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1716f060084065335630e2975b44a68791975ec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 08:27:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bfb87dabc8042f7a66e4b4c342b8d47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb74520419a6433299ef9232f8c15bf0 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca659deb31d143e2b85366981a096b27 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5db1da38219b410bb5b621bdb83095dd : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58b0d50d203247e6ba0c108c96277ea2 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26c45d3801a420e9c90951054f53433 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6ce0fd41d234587ab0b0242b024d1c9 : SUCCESS in 2m 58s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/0599bbee6ada4e9f8eb6807538797f95 : SUCCESS in 4m 01s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/468284a11b274ab4898b1bbb44e1fce8 : SUCCESS in 23m 27s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cde7648953543678a636cf51208f123 : SUCCESS in 29m 58s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eff609171b4e4a3c925392ff5fcd9d52 : SUCCESS in 18m 54s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce54285dbbfe4411b70952acefa376e8 : SUCCESS in 41m 00s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/57e0c0ed9ac343ab84c35c702cacbd60 : SUCCESS in 40m 35s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9601dd5c129942119cce19b262b9d57b : SUCCESS in 20m 11s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e883c52e9244776aaf5558d456fc7e2 : SUCCESS in 18m 01s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/7b5e5cf2384a48a7bdf4f31d5a15be51 : SUCCESS in 18m 20s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e711d2f3c8424126a26abbb00697f8a3 : SUCCESS in 39m 36s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c19511f53f24ebfb1e9099840cf3c41 : FAILURE in 21m 59s\n- bifrost-integration-tinyipa-debian-bullseye https://zuul.opendev.org/t/openstack/build/b9ba5edddc23413187492c8d19df42da : FAILURE in 4m 38s (non-voting)\n- bifrost-integration-tinyipa-centos-9 https://zuul.opendev.org/t/openstack/build/dad316b4649a40088dde59f8681cbac0 : FAILURE in 11m 14s (non-voting)\n- bifrost-integration-inspector-centos-9 https://zuul.opendev.org/t/openstack/build/b45b79f855a7425eac6309b171815957 : FAILURE in 11m 49s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/c147f179c37343c6b7e559e31bac7645 : FAILURE in 11m 51s (non-voting)\n- bifrost-integration-tinyipa-keystone-centos-9 https://zuul.opendev.org/t/openstack/build/cd844ed2b0c248aa860322428b044840 : FAILURE in 12m 13s (non-voting)\n- bifrost-upgrade-keystone-centos9 https://zuul.opendev.org/t/openstack/build/a136d94d0259463ca4725ecd652e95c9 : FAILURE in 1h 00m 21s (non-voting)\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc0502ae20a4445f8368981e2b67893d : FAILURE in 24m 06s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/1bd6a17070114b83875f0e2791e6a033 : FAILURE in 9m 07s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d34a8cd531b840f58dfef8e5d48f0b90 : SUCCESS in 37m 38s (non-voting)\n- bifrost-integration-dhcp-centos-9 https://zuul.opendev.org/t/openstack/build/5b1544d790964d89928db30c62b3bef0 : FAILURE in 9m 42s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/497d8df3e98e4af881f715a584cf885a : SUCCESS in 18m 47s (non-voting)\n- bifrost-integration-dibipa-debian-centos-9 https://zuul.opendev.org/t/openstack/build/d22634c2504247deb04a670e976ab670 : FAILURE in 13m 10s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c36d90f1f8c343d1a90579cd27da5fa4 : SUCCESS in 39m 42s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/b70c9dd123cb4e6c8fda0727138b7aaa : FAILURE in 3m 09s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/25c6c93e3ded49dbb6fcf1f1c42a9dc3 : SUCCESS in 41m 37s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/140438c80b66455f9fc9ec3f2e23d4b1 : SUCCESS in 34m 41s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/de0b7af80c01451087a032d9ad592767 : SUCCESS in 3m 16s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"a904c2197ecf128ad92e43efa1d33d1cbcec22f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-08-29 11:22:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"aeff26795ca748592c3fa2264ee5b0f76f1a0b0b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-29 11:46:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"aca476e3e80b1adc2c2ed3367a593286df3f3aca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-29 11:48:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"33d660241e083c3dd7f2d09498edc5dec7154e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:55:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c91b6e4434042f8833f5064ce520f14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae9b8c2f0b4f47f08a7158b0fac37fd6 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0f08cd6aee84a2499a3e7e03114ff2c : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9375404fde3042a590e81564e5c785fa : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/173482f99e634b108a3cebad91dcb7c1 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da3a8c7a76b94f45a5fe94f894af6b1f : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3dc8b08eaec41cf91394747eb73c2e8 : SUCCESS in 2m 07s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/88e6fdaa4a62428fa605b004bc8043b7 : SUCCESS in 8m 12s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf3de253287949f39248bfff045ce265 : SUCCESS in 42m 49s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4632b56129344a2d84ffffe06b8e7726 : SUCCESS in 19m 01s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e37c716b9ab4856aa781b1babb5848c : SUCCESS in 18m 35s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4fb2c8784cd48899563a0eeb6bd3daa : SUCCESS in 19m 57s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/c4baf8e0553842929ef2e103d6657e20 : SUCCESS in 21m 06s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a08591ee427a4948a1a1e3c9694f7f21 : SUCCESS in 19m 29s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47e9f5bc8d254279b779c281ed234bd8 : SUCCESS in 17m 40s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/e6f2a557aacc4002b07b0ed0b2b23200 : SUCCESS in 36m 06s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee089d2257974ab3bc6417237066b84a : SUCCESS in 21m 20s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfb075abd24e4aa289aae52f947bff5e : SUCCESS in 1h 19m 52s\n- bifrost-upgrade-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43c5c8d70df342e18df27a4ca4a64f61 : SUCCESS in 1h 27m 32s\n- bifrost-slurp-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fb72bf1fa7b4c628836958133852658 : FAILURE in 47m 34s (non-voting)\n- bifrost-integration-redfish-pxe-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8355732bb915491184523b5a28dd0d88 : FAILURE in 12m 43s (non-voting)\n- bifrost-integration-dhcp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efc0be5aae434bf3974c7bfb48dedaf7 : SUCCESS in 40m 06s (non-voting)\n- bifrost-integration-dibipa-debian-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b86594afe43242ad98f412bad4db7987 : SUCCESS in 36m 00s (non-voting)\n- bifrost-cli-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df984b4cc6234bc282307afbeb0875bd : SUCCESS in 41m 11s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/19d423b0815c4e23a6f9d964283fbdfc : SUCCESS in 35m 01s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/3b0dcd1a475b4bc1b2576eb5083e0712 : SUCCESS in 19m 47s (non-voting)\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/ee8f4739865f4e06aa4d2d547756277e : SUCCESS in 27m 09s (non-voting)\n- bifrost-tox-codespell https://zuul.opendev.org/t/openstack/build/90685720b5c946f69e6f50a03e04e7c6 : SUCCESS in 4m 53s (non-voting)\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":2},{"id":"8993eb16d92f29d448e4a216610f959e498a4674","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2025-08-29 13:33:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a3c917e136cd58920a7b85f2755764a0285c133","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-01 10:08:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"27acd91d42946e9d704c7b6339c5185fefd39f2b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-09-02 19:22:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"20004c790b1c15c72b1aa8a4e2a0a26ab6a835fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 19:23:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d8f0dd2696fc2b1600cf28cc0b86bc7a2548f8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 20:58:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e12229d8fad46959d64d5253c207f79\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2421462f386049d0a9e31920b2942a69 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9ef8f24521c4d8db0f3c023f3edc980 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88104a9e4f47426a9f2371b3d2baacd0 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b590bea2a234a9a9d1c40d7fa1a2e09 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0627ae4aeb6e4855bd17ca855cb04c57 : SUCCESS in 4m 57s\n- bifrost-tox-linters https://zuul.opendev.org/t/openstack/build/2ee18e7dd82f47bea9e092a62902ccd5 : SUCCESS in 4m 27s\n- bifrost-integration-tinyipa-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ace5c4d7525465cba5133aa40d29676 : SUCCESS in 37m 00s\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7362cb4a70f141f483e6af908cb2adf7 : SUCCESS in 31m 41s\n- bifrost-integration-tinyipa-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7fa73849f774eccb10440854afb87a9 : SUCCESS in 44m 54s\n- bifrost-integration-tinyipa-keystone-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcda5b89de16427389d2ef2caec22635 : SUCCESS in 28m 43s\n- bifrost-integration-tinyipa-debian-bookworm https://zuul.opendev.org/t/openstack/build/811666f1e20e407aacd59f2aa84c272c : SUCCESS in 40m 24s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcc433c194d042b09afdfca738e674a9 : SUCCESS in 41m 23s\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f87665609de74b6f846797ce985185b4 : SUCCESS in 39m 39s\n- bifrost-integration-redfish-vmedia-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/cbb974f21b054f0ea013ab5b5f35572e : SUCCESS in 18m 45s\n- bifrost-integration-inspector-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1f3922d50264e54851ac17d696ed036 : SUCCESS in 38m 13s\n- bifrost-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cca0df01451437cb63b99702072642f : SUCCESS in 43m 03s\n- bifrost-upgrade-keystone-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b5942b648164660ad6696b2857ddf49 : SUCCESS in 1h 30m 38s\n\nWarning:\n  Comments left for invalid file playbooks/roles/bifrost-ironic-install/defaults/main.yml","accounts_in_message":[],"_revision_number":2},{"id":"75ddd215b488baaf2f9098dc745989a304df80bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 20:58:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a7475e64ec3ae8f225df48d6f2075d5eb36a87a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 21:00:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a536380ee9ec46c8954713b29ebe0772\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01bb5cb86e0c4b45a07174222caa988e : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a81cd42f11fd4f3f92a63599b1ffac9b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33de042d4c890383bfab2de87fdb6627104c72e2","revisions":{"f912cc19e808d9796a88eb958ccbf553a7dcc8e8":{"kind":"REWORK","_number":1,"created":"2025-08-19 08:38:51.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/47/957847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/47/957847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/47/957847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/47/957847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/47/957847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/47/957847/1"}}},"commit":{"parents":[{"commit":"a43b5bd0f49a86c71894d8bb6f62b32e4e1c02ef","subject":"Move all CS9 jobs to non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/a43b5bd0f49a86c71894d8bb6f62b32e4e1c02ef"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-08-19 08:36:58.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-08-19 08:38:46.000000000","tz":120},"subject":"Deprecate support for Debian 11 Bullseye","message":"Deprecate support for Debian 11 Bullseye\n\nIronic has dropped support for Python 3.9 recently, the default\nPython version for Debian 11 Bullseye.\nWe move all the related jobs to non-voting for the time being,\nand deprecate its support, in preparation for complete removal.\n\nChange-Id: If3ff89ec4128fe2c12fa9c470da4f04163f5f667\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f912cc19e808d9796a88eb958ccbf553a7dcc8e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/f912cc19e808d9796a88eb958ccbf553a7dcc8e8"}]},"branch":"refs/heads/master"},"33de042d4c890383bfab2de87fdb6627104c72e2":{"kind":"REWORK","_number":2,"created":"2025-08-29 11:22:11.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/47/957847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/bifrost","ref":"refs/changes/47/957847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/47/957847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/47/957847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/bifrost refs/changes/47/957847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/bifrost refs/changes/47/957847/2"}}},"commit":{"parents":[{"commit":"86366ec4543b6c886020abbb062dcb4f4dd91124","subject":"Clarify testenv and install usage order in testenv.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/86366ec4543b6c886020abbb062dcb4f4dd91124"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-08-19 08:36:58.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2025-08-29 11:21:39.000000000","tz":120},"subject":"Recover the CI","message":"Recover the CI\n\n* Change the upgrade job to upgrade from the previous release.\n* Completely drop support for OS with Python 3.9.\n* Move jobs that only run on CS9 to Noble.\n\nChange-Id: If3ff89ec4128fe2c12fa9c470da4f04163f5f667\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/33de042d4c890383bfab2de87fdb6627104c72e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/bifrost/commit/33de042d4c890383bfab2de87fdb6627104c72e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
