)]}'
{"id":"openstack%2Fblazar-dashboard~755813","triplet_id":"openstack%2Fblazar-dashboard~stable%2Ftrain~I998d6929ad05d9b5bc4e07f27f3f9cbf2dd64c68","project":"openstack/blazar-dashboard","branch":"stable/train","topic":"bug/1895688-stable/train","hashtags":[],"change_id":"I998d6929ad05d9b5bc4e07f27f3f9cbf2dd64c68","subject":"Use json.loads instead of eval for JSON parsing","status":"MERGED","created":"2020-10-02 14:42:20.000000000","updated":"2020-10-02 17:24:39.000000000","submitted":"2020-10-02 17:23:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755813-1601659388399-5cec80e2","meta_rev_id":"51f5465ec886ce1b7c1167cb95703b6e7686ad46","_number":755813,"virtual_id_number":755813,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 17:23:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-02 17:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-02 17:00:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 17:23:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04faee0a8d30593737fb17e3153741c560babc87","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-02 14:42:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f81654034c94bbfa3f661a8635db8af348aafeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 14:53:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c48792564764b99b263bbb1160deb99 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4794f7fb4c4fcbbb832fa17665f3bf : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99e7db8beca546e6ae6e47372160b7b8 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/242a3a059ed944d89d4f66f3a76360c4 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad5f81f979ef4e7bb8cb868a0175d580 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc06bf419762407d92f4e234b9ca0bf8 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/061e76e317b84af381ab041f93769a63 : SUCCESS in 3m 49s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/307a5f3477bd4ab6b373d4539f23c0c5 : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5e04263901954b6e1b6a7b843af0cc252f95854b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-02 17:00:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec21ba1fc48c0d42e894a9cc8d25d07a59ac85ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:00:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"390812928ce174e209688c4f93ed9ad24b5aab16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:23:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a1b3d8c70fe44bc9ded03f72f35e522 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94901fa2fa9247278908deaeff93c4b3 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a38e47973674c8f82902be7db0e7428 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9bff96db06c4f389281683346d4bcb1 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32e682d07e49425d8322f6ea906e300f : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3250d881e1bf46dcbb5480058f71cbf0 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a533e9d5fa84f61825025d5e0cc5898 : SUCCESS in 3m 14s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/1b006903707a4cbe951ac93b08925f68 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"bd983966fa1c520fa395cb62c5739a632492a961","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:23:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c4dae9eb0aea2ad2cbe04dd7621e8e7806f0b0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:24:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead74fe5ea5a4b74a2bc8b26980d07f3 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df51e07efa6141bba47f11b453e8419a : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63e9c5d25617467016eea1dff0a34803c86b0953","revisions":{"63e9c5d25617467016eea1dff0a34803c86b0953":{"kind":"REWORK","_number":1,"created":"2020-10-02 14:42:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/755813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-dashboard","ref":"refs/changes/13/755813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-dashboard refs/changes/13/755813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-dashboard refs/changes/13/755813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-dashboard refs/changes/13/755813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-dashboard refs/changes/13/755813/1"}}},"commit":{"parents":[{"commit":"e39e27a12f26362f34c2d6b5b29a5ce4127fcdf2","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-dashboard/commit/e39e27a12f26362f34c2d6b5b29a5ce4127fcdf2"}]}],"author":{"name":"Lukas Euler","email":"lukas@positive-security.io","date":"2020-09-15 13:25:40.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-01 09:38:07.000000000","tz":120},"subject":"Use json.loads instead of eval for JSON parsing","message":"Use json.loads instead of eval for JSON parsing\n\nAlso fixed error messages.\n\nChange-Id: I998d6929ad05d9b5bc4e07f27f3f9cbf2dd64c68\nCloses-Bug: #1895688\n(cherry picked from commit 33c58438abf8221291d264db26a061279d4f22c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-dashboard/commit/63e9c5d25617467016eea1dff0a34803c86b0953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-dashboard/commit/63e9c5d25617467016eea1dff0a34803c86b0953"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
