)]}'
{"id":"openstack%2Fblazar-nova~663949","triplet_id":"openstack%2Fblazar-nova~master~I07b3411044c88c3bf51db2515208902a6857817b","project":"openstack/blazar-nova","branch":"master","topic":"sphinx-requirements","hashtags":[],"change_id":"I07b3411044c88c3bf51db2515208902a6857817b","subject":"Update sphinx requirements","status":"MERGED","created":"2019-06-07 14:04:33.000000000","updated":"2019-06-18 09:31:50.000000000","submitted":"2019-06-18 09:31:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663949-1560850310982-d473cd44","meta_rev_id":"591f74236a2d182ef5395781fff59a2f2abebf74","_number":663949,"virtual_id_number":663949,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-06-18 09:31:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":2,"date":"2019-06-18 09:18:21.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-18 09:12:15.000000000","_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":1,"date":"2019-06-18 09:18:21.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 09:12:15.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2019-06-18 09:18:21.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-06-18 09:31:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8642be690cebe9d68a882009768c0dfb6e8d2dd9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-07 14:04:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3389dd50b42e885d7712514bbaab1f3242a3cb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:20:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/49/663949/1/check/requirements-check/7d62a65/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/49/663949/1/check/openstack-tox-pep8/4089f56/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/49/663949/1/check/openstack-tox-py27/8850987/ : SUCCESS in 6m 09s\n- openstack-tox-py36 http://logs.openstack.org/49/663949/1/check/openstack-tox-py36/457b1bf/ : SUCCESS in 6m 33s\n- openstack-tox-py37 http://logs.openstack.org/49/663949/1/check/openstack-tox-py37/77dd1d3/ : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/663949/1/check/openstack-tox-lower-constraints/b699990/ : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":1},{"id":"81b96067f9fc99a57e06306c8f3283b31b57cf24","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2019-06-18 09:12:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2fd77911c386bf777b6e0855dbbca2c566843fb","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-06-18 09:18:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb47daf5170ee17620be7e9243c719b6df71e26d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:18:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"731563992d66f7b285346d19790151026ddc7da7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:31:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a9e390b3f2c28a7bbc09b228ff8580914d2709e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:31:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/49/663949/1/gate/requirements-check/af2f425/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/49/663949/1/gate/openstack-tox-pep8/322915a/ : SUCCESS in 6m 45s\n- openstack-tox-py27 http://logs.openstack.org/49/663949/1/gate/openstack-tox-py27/944b653/ : SUCCESS in 7m 03s\n- openstack-tox-py36 http://logs.openstack.org/49/663949/1/gate/openstack-tox-py36/ec98261/ : SUCCESS in 6m 47s\n- openstack-tox-py37 http://logs.openstack.org/49/663949/1/gate/openstack-tox-py37/1563187/ : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/663949/1/gate/openstack-tox-lower-constraints/9fde218/ : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a8f8927b4328ac33ee4b566aaab4b0761575035","revisions":{"4a8f8927b4328ac33ee4b566aaab4b0761575035":{"kind":"REWORK","_number":1,"created":"2019-06-07 14:04:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/663949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/49/663949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/49/663949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/49/663949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/49/663949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/49/663949/1"}}},"commit":{"parents":[{"commit":"184b3598a4aa1c6ca422e7d16330e34b028fb10f","subject":"Use opendev repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/184b3598a4aa1c6ca422e7d16330e34b028fb10f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-07 14:04:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-07 14:04:21.000000000","tz":60},"subject":"Update sphinx requirements","message":"Update sphinx requirements\n\nFollowing recent changes to global requirements [1], we need\nto update sphinx requirements locally. This should fix the\nrequirements-check job.\n\n[1] If558f184c959e4b63b56dec3ca1571d1034cfe5c\n\nChange-Id: I07b3411044c88c3bf51db2515208902a6857817b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/4a8f8927b4328ac33ee4b566aaab4b0761575035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/4a8f8927b4328ac33ee4b566aaab4b0761575035"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
