)]}'
{"id":"openstack%2Fblazar-nova~768975","triplet_id":"openstack%2Fblazar-nova~master~I53ee9b459c595d80768650ba20665bc633e90b71","project":"openstack/blazar-nova","branch":"master","hashtags":[],"change_id":"I53ee9b459c595d80768650ba20665bc633e90b71","subject":"remove py37","status":"ABANDONED","created":"2021-01-03 07:21:21.000000000","updated":"2021-01-25 09:13:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0010ab13350385408d50a16bd7c409f2c19f7a0c","_number":768975,"virtual_id_number":768975,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-03 07:38:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-01-12 01:33:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":-2,"date":"2021-01-05 21:13:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 07:38:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-05 21:13:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-01-12 01:33:19.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"10b0f993587e71eca04ac0c4b90201086bca62ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-01-03 07:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c202a45aebcedb7135cab5549377a0ad9987a7b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-01-03 07:27:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8c263b6f6e2f9f79897b2d76ef9ef6e751b1074d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 07:38:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/83f46c34d3314ec784dd99c43a5fc1e2 : SUCCESS in 2m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc1b7f547a894744aa61140082f4c039 : FAILURE in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b74562caa80c4967b4921be8160b32d8 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/019083d3b5874c00b6e7b43567c0a1d3 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ea0a4e1db54adcb59668f9c7d50d8c : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55343f74b06a423cab51d952c10ac6d4 : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eafab16ad60f161538d6c054db8a7bf303d26023","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-01-03 07:55:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a7aa35684a6c0b856a2ea14c548f4fab1c1e7188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 08:03:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e34f26f39644fb192708df8ab5efbfd : SUCCESS in 1m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5457d8e2dd54fa1bd05314c72b3ec3d : FAILURE in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1445ed7e8e474528ae31dd356f4fa671 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/243c339921d44f94897d5a7f8e4b40a0 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f1a266504d346b0be36f8892efc20b2 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03fe84cc9ef34494af13b66706e99fd2 : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9150ca714768179f44a7a6ac51a3e6bbc244ec2b","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-01-04 00:41:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a3a0d8817bbb00f3a71df0e017f6a12e4b014210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 00:52:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b76b4f296034b5fb0d03466803a1fef : SUCCESS in 1m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f17845d18814eb38e56823050c33e18 : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5ba3211f074bd9822ee59818e8300d : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4da032b04b9e4871afbf48a957dc6192 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/626cfe31b15b4e99bbe6992f1e1e7553 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36bbfc8f0e9f41a5981faa12c3fc1c10 : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cddfb16709aadb7328f48f90a79a59622cfc96cc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-05 21:13:15.000000000","message":"Patch Set 2: Code-Review-2\n\nAs Akihiro Motoki says in https://review.opendev.org/c/openstack/horizon/+/769237:\n\nThis looks like a misinterpretation of the PTI. Testing python 3.6 and python 3.8 is almost enough to support python 3.7, so the PTI recommends to test python 3.6 and 3.8 only.","accounts_in_message":[],"_revision_number":2},{"id":"bfd3be0584b808c344c97f9123713786662530c3","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-12 01:33:19.000000000","message":"Patch Set 2: Code-Review-1\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-January/019649.html","accounts_in_message":[],"_revision_number":2},{"id":"0010ab13350385408d50a16bd7c409f2c19f7a0c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-25 09:13:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff07a94a2425b411d49d70a7052f94a37b77dc7e","revisions":{"68a00005b056fd01ed9e3ed4c7008dfc08ba2c93":{"kind":"REWORK","_number":1,"created":"2021-01-03 07:21:21.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/75/768975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/75/768975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/75/768975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/75/768975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/75/768975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/75/768975/1"}}},"commit":{"parents":[{"commit":"d399fe717b9185da06d9f3dd42117294aaf3f2cc","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/d399fe717b9185da06d9f3dd42117294aaf3f2cc"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-03 07:20:21.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-03 07:21:09.000000000","tz":480},"subject":"remove py37","message":"remove py37\n\nChange-Id: I53ee9b459c595d80768650ba20665bc633e90b71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/68a00005b056fd01ed9e3ed4c7008dfc08ba2c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/68a00005b056fd01ed9e3ed4c7008dfc08ba2c93"}]},"branch":"refs/heads/master"},"ff07a94a2425b411d49d70a7052f94a37b77dc7e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-03 07:27:23.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/75/768975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/75/768975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/75/768975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/75/768975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/75/768975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/75/768975/2"}}},"commit":{"parents":[{"commit":"d399fe717b9185da06d9f3dd42117294aaf3f2cc","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/d399fe717b9185da06d9f3dd42117294aaf3f2cc"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-03 07:20:21.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-03 07:26:47.000000000","tz":480},"subject":"remove py37","message":"remove py37\n\nRemove python3.7 from setup.cfg, since Wallaby\u0027s python\nsupported runtimes are python 3.6 and python 3.8[1]:\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/wallaby.html\n\nChange-Id: I53ee9b459c595d80768650ba20665bc633e90b71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/ff07a94a2425b411d49d70a7052f94a37b77dc7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/ff07a94a2425b411d49d70a7052f94a37b77dc7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
