)]}'
{"id":"openstack%2Fblazar-nova~891784","triplet_id":"openstack%2Fblazar-nova~master~Iba090f5e9654e39db6e2004d79d254bcad33071b","project":"openstack/blazar-nova","branch":"master","topic":"release-health-check-cwi-bobcat","hashtags":[],"change_id":"Iba090f5e9654e39db6e2004d79d254bcad33071b","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2023-08-17 14:55:32.000000000","updated":"2023-08-18 07:42:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f9d7fc74a60c99736e0917f3dc432246b06b5724","_number":891784,"virtual_id_number":891784,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-17 15:18:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 15:18:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa8396fae41feb9f7d35547dcbf7e7b383e99c5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-17 14:55:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43c1346a05d54588e415aaa2dc7829eb2af5f2f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 15:18:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/500ffba8f46c4c3f824dbf8829a07753\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc4121769cdd404391fa77f05bf7078c : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbd245dd4039436c83993725c744c1f6 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e795f4ffbddf41e78f0bd58f7a819bfc : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f9d7fc74a60c99736e0917f3dc432246b06b5724","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-18 07:42:23.000000000","message":"Abandoned\n\nGate looks healthy, abandoning.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66fca654fb212522eca9cbaf3ff67636a6ab003b","revisions":{"66fca654fb212522eca9cbaf3ff67636a6ab003b":{"kind":"REWORK","_number":1,"created":"2023-08-17 14:55:32.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/84/891784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/84/891784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/84/891784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/84/891784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/84/891784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/84/891784/1"}}},"commit":{"parents":[{"commit":"021780c99d85eeb4cf5ba653bd1c1501e2c88130","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/021780c99d85eeb4cf5ba653bd1c1501e2c88130"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-17 14:55:19.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-08-17 14:55:29.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: Iba090f5e9654e39db6e2004d79d254bcad33071b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/66fca654fb212522eca9cbaf3ff67636a6ab003b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/66fca654fb212522eca9cbaf3ff67636a6ab003b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
