)]}'
{"id":"openstack%2Fblazar-nova~987002","triplet_id":"openstack%2Fblazar-nova~master~Iee71fab48cdcba8c3f96db9d49f27557a8884826","project":"openstack/blazar-nova","branch":"master","attention_set":{},"removed_from_attention_set":{"28356":{"account":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"last_update":"2026-05-27 05:29:17.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-27 05:03:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-27 05:29:17.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-05-14 13:55:34.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Iee71fab48cdcba8c3f96db9d49f27557a8884826","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2026-05-02 14:52:51.000000000","updated":"2026-05-27 05:29:17.000000000","submitted":"2026-05-27 05:29:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987002","meta_rev_id":"02cf7208e64d161fa48a3025c5de39a7749b219e","_number":987002,"virtual_id_number":987002,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 05:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-14 15:52:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":2,"date":"2026-05-14 13:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-14 15:52:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-02 15:16:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 19:18:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"state":"REVIEWER"},{"updated":"2026-05-11 19:18:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-05-11 19:18:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-05-14 15:52:13.000000000","updated_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"}],"messages":[{"id":"825f66a0fc95a6c677e9b79e2c887676636d8f08","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 14:52:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc942179e22093a5ef8363deaf00435bf7cd596a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 15:01:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"35469930f236151734d2cbf66cf61f92a82e9294","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 15:01:22.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"acbc4c52dac7a23eefbdb749abc43d479aa6e8b3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 15:05:04.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7fe65f74b2be046562a698d860d29f613c86dd99","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 15:05:07.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"7114736c3f7a6cfe7a0535e1490ae7c6fd52f1b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 15:16:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0628137c9db8452aae4a985a7c31006c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36351480b8cc47569831197f6e98f345 : SUCCESS in 1m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b8ceb263e594acea1eb0fec2a6d044b : SUCCESS in 2m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f3c60e9207c4496b99634f1e1b88e1c : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3d27b9349154c9695dcdb3f1b4e027f : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18db195f5be748c68c87f8344d763b63 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d957f3025b3c44cda2140d3d8d2b598b : SUCCESS in 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9081a178e70b0b8b59af50242c0888e22288a9e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 20:45:27.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b4b6d1f70230cb9cf69fffbd32a04f952eefdf27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 20:52:52.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"4962105b0fa6140e2dcbd3a5a872f939e89d9023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 20:53:25.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"b1127e9b53eddf74e66214bfb459055c7ae62756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 21:11:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd4a6bea27fe4c2c8ec6e4120d1cd550\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/53088bc2b87c498bbab97c4440c6c6ba : SUCCESS in 2m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f04f0450d8a0464f893a81d60c3bf825 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59f6fffcafb14af7b9ba541cd30bcce7 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c4409d936644cbbba2fb2378b6a67a8 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2134644ca9e6413ba2757ec679a951df : SUCCESS in 6m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/18ae1dcdb63048668bc0f830ccd659b8 : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"415430d474a63fa86054aa47b5747feab217a739","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-14 13:55:34.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"df8072f347602516340d3abc97984aa069acb174","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2026-05-14 15:52:13.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"bd00de99e406b4e154816348eab273346265c8f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:52:49.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"45d6dd1958acc7ddec07b03610c7b04685f323a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 16:04:45.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/blazar-nova/+/987001 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"8104dd4c48c3d9cdc91349ac07becd1ab7275dff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-27 05:03:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ce8250450ad754d98ff60cb9b1ef714c7ed63143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 05:18:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c2b382053641918adf57977a863e12\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f58dac8da7b4ba6862eb6a30bda070a : SUCCESS in 4m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ee333e3f119426d9907a4256cbdc256 : SUCCESS in 1m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c96ff13db84eae9b685e447969c4a7 : SUCCESS in 2m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4cbb8994025e426dae6e3bfac0e8ef4b : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b0570ac2b1f4ed3bd025ddba776e8f2 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0cd1f368fc8a4aefa6a66e2a93f2f5c6 : SUCCESS in 7m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ceb6e2dc7604b4ebf0cb954bb92e9d0 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":7},{"id":"ed5dee38f1baf2be41e4e5f949e71c6754214d72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 05:20:06.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ff09da9ac8f636064796dd20923bb637dd0c44d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 05:29:17.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c88ed112d774c70bd3a7ce65ee816a5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/229be732cffc474bb67591b1424ca8b5 : SUCCESS in 2m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/97e8aa52bf234e089c36f7ca20b92c54 : SUCCESS in 1m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54fa2a3ca8ed43f7816a71cb4fb718a8 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d8909ecdaf64aa59d47b435d011a270 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73b6b696db554ce796fd2f5073f8643f : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53ff2d27a5a64c5fbbf04e01e7ce14d9 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":7},{"id":"02cf7208e64d161fa48a3025c5de39a7749b219e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 05:29:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"134331f39a4491201569445505d97e380a3ce63b","revisions":{"4bc07b1aadc8772833edae5ee905a97352ffa076":{"kind":"REWORK","_number":1,"created":"2026-05-02 14:52:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/987002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/02/987002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/1"}}},"commit":{"parents":[{"commit":"9f5f9c53b2cf24197eed046b8e6c3c6d156fdd8a","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/9f5f9c53b2cf24197eed046b8e6c3c6d156fdd8a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 14:52:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 14:52:51.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Iee71fab48cdcba8c3f96db9d49f27557a8884826\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/4bc07b1aadc8772833edae5ee905a97352ffa076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/4bc07b1aadc8772833edae5ee905a97352ffa076"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"99e87f6dafcee52d241afde766ff95d3c8fd768d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-02 15:01:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/987002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/02/987002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/2"}}},"commit":{"parents":[{"commit":"9f5f9c53b2cf24197eed046b8e6c3c6d156fdd8a","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/9f5f9c53b2cf24197eed046b8e6c3c6d156fdd8a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 14:52:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 15:01:07.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iee71fab48cdcba8c3f96db9d49f27557a8884826\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/99e87f6dafcee52d241afde766ff95d3c8fd768d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/99e87f6dafcee52d241afde766ff95d3c8fd768d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f021c90f082fea42dabb8916680382154a7b5a8c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-02 15:01:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/987002/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/02/987002/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/3"}}},"commit":{"parents":[{"commit":"f3a1e33c94d4041e68758cbd98711f326388f9bb","subject":"Remove pbr.hooks.setup_hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/f3a1e33c94d4041e68758cbd98711f326388f9bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 14:52:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 15:01:22.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iee71fab48cdcba8c3f96db9d49f27557a8884826\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/f021c90f082fea42dabb8916680382154a7b5a8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/f021c90f082fea42dabb8916680382154a7b5a8c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"99e87f6dafcee52d241afde766ff95d3c8fd768d","theirs":"f3a1e33c94d4041e68758cbd98711f326388f9bb","contains_conflicts":false}},"e1b18772d6031d1d7883ad8ab7a2483e4bdeeb15":{"kind":"REWORK","_number":4,"created":"2026-05-02 15:05:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/987002/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/02/987002/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/4"}}},"commit":{"parents":[{"commit":"f3a1e33c94d4041e68758cbd98711f326388f9bb","subject":"Remove pbr.hooks.setup_hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/f3a1e33c94d4041e68758cbd98711f326388f9bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 14:52:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 15:05:04.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iee71fab48cdcba8c3f96db9d49f27557a8884826\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/e1b18772d6031d1d7883ad8ab7a2483e4bdeeb15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/e1b18772d6031d1d7883ad8ab7a2483e4bdeeb15"}]},"branch":"refs/heads/master"},"5f249d23bb0105b9381bc07e1ad9edac2c486664":{"kind":"REWORK","_number":5,"created":"2026-05-04 20:45:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/987002/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/02/987002/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/5"}}},"commit":{"parents":[{"commit":"f3a1e33c94d4041e68758cbd98711f326388f9bb","subject":"Remove pbr.hooks.setup_hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/f3a1e33c94d4041e68758cbd98711f326388f9bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 14:52:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 20:45:27.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iee71fab48cdcba8c3f96db9d49f27557a8884826\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/5f249d23bb0105b9381bc07e1ad9edac2c486664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/5f249d23bb0105b9381bc07e1ad9edac2c486664"}]},"branch":"refs/heads/master"},"8f7406b2e19e033f8fe21ae04b45a284e7e6a908":{"kind":"REWORK","_number":6,"created":"2026-05-04 20:52:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/987002/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/02/987002/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/6"}}},"commit":{"parents":[{"commit":"f3a1e33c94d4041e68758cbd98711f326388f9bb","subject":"Remove pbr.hooks.setup_hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/f3a1e33c94d4041e68758cbd98711f326388f9bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 14:52:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 20:52:51.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iee71fab48cdcba8c3f96db9d49f27557a8884826\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/8f7406b2e19e033f8fe21ae04b45a284e7e6a908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/8f7406b2e19e033f8fe21ae04b45a284e7e6a908"}]},"branch":"refs/heads/master"},"134331f39a4491201569445505d97e380a3ce63b":{"kind":"REWORK","_number":7,"created":"2026-05-04 20:53:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/987002/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-nova","ref":"refs/changes/02/987002/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-nova refs/changes/02/987002/7"}}},"commit":{"parents":[{"commit":"f3a1e33c94d4041e68758cbd98711f326388f9bb","subject":"Remove pbr.hooks.setup_hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/f3a1e33c94d4041e68758cbd98711f326388f9bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 14:52:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 20:53:25.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iee71fab48cdcba8c3f96db9d49f27557a8884826\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/134331f39a4491201569445505d97e380a3ce63b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-nova/commit/134331f39a4491201569445505d97e380a3ce63b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
