)]}'
{"id":"openstack%2Fblazar-tempest-plugin~691551","triplet_id":"openstack%2Fblazar-tempest-plugin~master~Ia18f527ccacc14630a3ec4a2335c99a654b7fea0","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Ia18f527ccacc14630a3ec4a2335c99a654b7fea0","subject":"Stop testing with py2","status":"MERGED","created":"2019-10-28 02:18:12.000000000","updated":"2020-01-27 23:09:32.000000000","submitted":"2020-01-27 23:07:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691551-1580166444060-46a14a57","meta_rev_id":"d4cc7fa4dca507bc11ba9529c95b3dcd07588fdd","_number":691551,"virtual_id_number":691551,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 23:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-22 16:29:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-01-27 22:13:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-01-27 22:13:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 10:04:33.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2020-01-22 16:29:35.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-01-27 22:13:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-01-27 23:07:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb19d9d248e1156323718b916031e8f0e34a5180","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-10-28 02:18:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50d654e5e7897472d52ce1ee8b273f9a69e5c2a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 03:01:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f2e2656faddd4d8bb096f4daca9181ff : SUCCESS in 1m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d124b5fe7534f039f1ebcd9ce3dcf12 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d581327fb2f44c0b8d2b71175fb32321 : SUCCESS in 5m 10s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/c8a42688716f4f8c8fd789f2eea7ce33 : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c5b7f2711631ccd92e21a1959a4ffe0c660dda47","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-03 06:05:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f944e168f6f4dd2832cedc254c3f27410e600cd0","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-03 06:05:58.000000000","message":"Topic changed from drop-python2 to drop-py27-support","accounts_in_message":[],"_revision_number":2},{"id":"9d3dabc4435114a38edea6d6e1c88b7f3ff815fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 06:47:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a36eea73f8554921bb36f2007832364e : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ea9caa0a2a643c4927be03fc18ca0c3 : SUCCESS in 4m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6c82c858f72c4af1bc998236a33f0b68 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fed5f5097ff4feabe29637acec673e5 : SUCCESS in 7m 29s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/1adc8496a5354851ba3c9824b139cc95 : SUCCESS in 39m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e26f83278c7fb7f2b163ac2e47c9c1ef31417629","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-12-05 10:04:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6db2bd6d426aa237fc4f1e7848ddcb316cc28724","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-05 20:47:40.000000000","message":"Patch Set 2:\n\nwe need to wait till phase-2 as per discussed schedule- https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#schedule\n\nAlso we need to make sure old py2 jobs run on py2 till stable/train. I am going to start the audit of all tempest plugins by next week or so.","accounts_in_message":[],"_revision_number":2},{"id":"409571a789c26827334eccf9966e48a290a0d99d","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-06 04:13:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7420c5849877373c76a1b7f3c6dbe7b6e41a9312","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-06 04:13:27.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"dbce20d581c100edbaf00b42ce6484615a4e7bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 04:50:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cda47bce1cc14566b7a8a20469b758ae : SUCCESS in 1m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473217d5bcba4f0aa52cd26685e3333f : SUCCESS in 2m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf04873e891543c1b79d6110d0255c9b : SUCCESS in 1m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c96089204fd41fa9b31173ab75d527b : SUCCESS in 4m 52s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/015a7879cba94133adf2687496580474 : SUCCESS in 35m 52s","accounts_in_message":[],"_revision_number":3},{"id":"6c64f4c91cc96d294d949aad1bead1a7c1a631be","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-21 09:24:06.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"eb54afc082df4003b41ae1c85813a88220153509","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-21 09:34:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"94dd8c0503e4594b9e4e4b124660239f98685713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 10:28:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3936a23ca90b43f9a6b2d40606d34db1 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7875400ade324beeb2903c39f54af922 : SUCCESS in 5m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8224676ead0d479e8a7e17515113450f : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180fe51d1d1a4dcbaec04432d366b5ab : SUCCESS in 6m 23s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/eabfd60eac4a40cb993225f196a91dc7 : SUCCESS in 46m 51s","accounts_in_message":[],"_revision_number":4},{"id":"1e29745331a76285ac669cca684a2ad3b9fade13","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-21 11:21:57.000000000","message":"Patch Set 4:\n\n@gmann: Is it safe to merge this patch now, with regards to stable branch testing?","accounts_in_message":[],"_revision_number":4},{"id":"17e72c57fd619bae59e0fa30dc4b74ca82587a1c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-21 18:00:03.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nyeah, this is good to go. this plugin does not deifne the zuul jobs used for stable testing so we do not need to adjust it for stable vs master testing. \n\nOn stable testing, I am trying to run tempest and plugins on venv with 3.6 always, that will solve the overall issues if occur any - https://review.opendev.org/#/c/703476/","accounts_in_message":[],"_revision_number":4},{"id":"d846ed9df3f365876c879ceeb1d56ed8da725b4a","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-22 05:52:29.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"099562f92bb8d8ef36695538b7f11f296147cb06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 06:05:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fada1d32c0a94f1daa7d8192ece8dc4e : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b53f9e17514020ba8602b40332bb6f : SUCCESS in 3m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2958d68cb50644189363ccd0a0e266da : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c7d91ec6cf456c822226e2560d32e5 : SUCCESS in 5m 30s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/09dd7655a65d48518bd46a7c9ebe1b8f : FAILURE in 11m 24s","accounts_in_message":[],"_revision_number":5},{"id":"82398ef5475352931f2be699c6a55340bfb73093","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-22 16:29:35.000000000","message":"Patch Set 5: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"7523cfa2d8da8df6a14f93fe5a88a129f81e2514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 16:52:46.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1882b66454894406b2dcbaf6c7dc2430 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d3b4fd368b94972bcad5b681e23a51e : SUCCESS in 5m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7fdb138b3ce94c6aab4c7050a1643f9e : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79138a48bf5f44fa8ecf04e1f9feabd7 : SUCCESS in 6m 32s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/4d20a8d570e44d3693ffdde2ccdceaea : FAILURE in 16m 54s","accounts_in_message":[],"_revision_number":5},{"id":"b1c38b3978526a7c351505758839b030e45fa11e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-27 21:22:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d13138329d5e847d62c44d54e0c78e1f6e1ecc34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 22:05:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d2c33fd4ead84e5c8424030ed709e708 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277a81a49bd84565aee33b908083b7c8 : SUCCESS in 7m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5872e4dae6354f54969ec02683f23410 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd81caa497a4d4fa8cb719b2c970fdf : SUCCESS in 9m 16s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/b92fb8f1bf6648b99abf8361de19fd30 : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":5},{"id":"adc3728402f56a23c4876f6db1bc3f106bd1deb7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-27 22:13:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4ccca644667a8020698acdc2401ea8b891be7b77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 22:14:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"be152d955966ae889d7012903eb881ae37451b51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 23:07:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ef43d0c753e64f46969cdde724d1d635 : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee97eddbc084a83a2003924e6f9179c : SUCCESS in 3m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/69c12c6942e8498db06a2b4981ca47d5 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/669600fc43e74478bede08528271d831 : SUCCESS in 8m 23s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/3f1bacbdf4984b4e8061aee5d1225f11 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":5},{"id":"a7e06d90cc8198f62d34c782ef9e8b955e3c633a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 23:07:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"d4cc7fa4dca507bc11ba9529c95b3dcd07588fdd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 23:09:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d67a55e8d2435490eb9193b3c4b12c : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"31b690c1f768774177c57ab856eedb675d4af1f6","revisions":{"707a78cf1e34b5e9fe4c9ceb8fc957dfb4cf22ef":{"kind":"REWORK","_number":1,"created":"2019-10-28 02:18:12.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/51/691551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/51/691551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/1"}}},"commit":{"parents":[{"commit":"5e3f2736fea67a96316315ad0ed2d2730fb47432","subject":"Generate PDF documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/5e3f2736fea67a96316315ad0ed2d2730fb47432"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 02:17:03.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 02:17:58.000000000","tz":0},"subject":"Stop testing with py2","message":"Stop testing with py2\n\nPython 2 is no longer supported. Python 3 is required.\nThis patch stops testing with Python 2.\n\nChange-Id: Ia18f527ccacc14630a3ec4a2335c99a654b7fea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/707a78cf1e34b5e9fe4c9ceb8fc957dfb4cf22ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/707a78cf1e34b5e9fe4c9ceb8fc957dfb4cf22ef"}]},"branch":"refs/heads/master"},"8c7632bba8b3d6c7ca3fa0e6821773d5d81db2be":{"kind":"REWORK","_number":2,"created":"2019-12-03 06:05:43.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/51/691551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/51/691551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/2"}}},"commit":{"parents":[{"commit":"5e3f2736fea67a96316315ad0ed2d2730fb47432","subject":"Generate PDF documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/5e3f2736fea67a96316315ad0ed2d2730fb47432"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 02:17:03.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 15:14:38.000000000","tz":0},"subject":"Stop testing with py2","message":"Stop testing with py2\n\nPython 2 is no longer supported. Python 3 is required.\nThis patch stops testing with Python 2.\n\nChange-Id: Ia18f527ccacc14630a3ec4a2335c99a654b7fea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/8c7632bba8b3d6c7ca3fa0e6821773d5d81db2be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/8c7632bba8b3d6c7ca3fa0e6821773d5d81db2be"}]},"branch":"refs/heads/master"},"f89232671aec826422874f3d5aa7dcaa5fbf67c8":{"kind":"REWORK","_number":3,"created":"2019-12-06 04:13:12.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/51/691551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/51/691551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/3"}}},"commit":{"parents":[{"commit":"5e3f2736fea67a96316315ad0ed2d2730fb47432","subject":"Generate PDF documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/5e3f2736fea67a96316315ad0ed2d2730fb47432"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 02:17:03.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-12-06 04:13:08.000000000","tz":0},"subject":"Stop testing with py2","message":"Stop testing with py2\n\nPython 2 is no longer supported. Python 3 is required.\nThis patch stops testing with Python 2.\n\nChange-Id: Ia18f527ccacc14630a3ec4a2335c99a654b7fea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/f89232671aec826422874f3d5aa7dcaa5fbf67c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/f89232671aec826422874f3d5aa7dcaa5fbf67c8"}]},"branch":"refs/heads/master"},"020c5c6be0b8c08863923d3a3802a41b635dd294":{"kind":"REWORK","_number":4,"created":"2020-01-21 09:34:10.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/51/691551/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/51/691551/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/4"}}},"commit":{"parents":[{"commit":"5e3f2736fea67a96316315ad0ed2d2730fb47432","subject":"Generate PDF documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/5e3f2736fea67a96316315ad0ed2d2730fb47432"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 02:17:03.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-01-20 22:34:34.000000000","tz":0},"subject":"Stop testing with py2","message":"Stop testing with py2\n\nPython 2 is no longer supported. Python 3 is required.\nThis patch stops testing with Python 2.\n\nChange-Id: Ia18f527ccacc14630a3ec4a2335c99a654b7fea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/020c5c6be0b8c08863923d3a3802a41b635dd294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/020c5c6be0b8c08863923d3a3802a41b635dd294"}]},"branch":"refs/heads/master"},"31b690c1f768774177c57ab856eedb675d4af1f6":{"kind":"REWORK","_number":5,"created":"2020-01-22 05:52:29.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/51/691551/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/51/691551/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/51/691551/5"}}},"commit":{"parents":[{"commit":"5e3f2736fea67a96316315ad0ed2d2730fb47432","subject":"Generate PDF documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/5e3f2736fea67a96316315ad0ed2d2730fb47432"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-28 02:17:03.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-01-22 05:52:07.000000000","tz":0},"subject":"Stop testing with py2","message":"Stop testing with py2\n\nPython 2 is no longer supported. Python 3 is required.\nThis patch stops testing with Python 2.\n\nChange-Id: Ia18f527ccacc14630a3ec4a2335c99a654b7fea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/31b690c1f768774177c57ab856eedb675d4af1f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/31b690c1f768774177c57ab856eedb675d4af1f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
