)]}'
{"id":"openstack%2Fblazar-tempest-plugin~736454","triplet_id":"openstack%2Fblazar-tempest-plugin~master~Ic8b2d159a538d31f7f58a21afe7dac533dfd9314","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"optimization/pip-no-cache-dir","hashtags":[],"change_id":"Ic8b2d159a538d31f7f58a21afe7dac533dfd9314","subject":"optimize size and time using \"--no-cache-dir\"","status":"ABANDONED","created":"2020-06-18 06:29:52.000000000","updated":"2020-06-18 12:23:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bb19bca9a10cad1a90e37687eaed433a7e0f7146","_number":736454,"virtual_id_number":736454,"owner":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-18 07:11:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 07:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e1114772bc3c90b43489185ab7857aabd98b74f","author":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"date":"2020-06-18 06:29:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c8d76075d48857972af90ea5a1a3335c31b6770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:11:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b89000a927a4994a3100f482e73c454 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6344413650c4652926cf2e18e832fec : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5befb51a1105dc59446c92d2cdda630771e716ba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 12:23:34.000000000","message":"Abandoned\n\nSee http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015515.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29fe73e33415cf2b77a886eab650e23163e5571f","revisions":{"29fe73e33415cf2b77a886eab650e23163e5571f":{"kind":"REWORK","_number":1,"created":"2020-06-18 06:29:52.000000000","uploader":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"ref":"refs/changes/54/736454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/54/736454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/54/736454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/54/736454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/54/736454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/54/736454/1"}}},"commit":{"parents":[{"commit":"9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f"}]}],"author":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-06-18 04:39:32.000000000","tz":330},"committer":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-06-18 06:29:45.000000000","tz":330},"subject":"optimize size and time using \"--no-cache-dir\"","message":"optimize size and time using \"--no-cache-dir\"\n\nUsing \"--no-cache-dir\" flag in pip install ,make sure dowloaded packages\nby pip don\u0027t cached on system . This is a best practise which make sure\nto fetch ftom repo instead of using local cached one . Further , in case\nof Docker Containers , by restricing caching , we can reduce image size.\nIn term of stats , it depends upon the number of python packages\nmultiplied by their respective size . e.g for heavy packages with a lot\nof dependencies it reduce a lot by don\u0027t caching pip packages.\n\nFurther , more detail information can be found at\n\nhttps://medium.com/sciforce/strategies-of-docker-images-optimization-2ca9cc5719b6\n\nChange-Id: Ic8b2d159a538d31f7f58a21afe7dac533dfd9314\nSigned-off-by: Pratik Raj \u003crajpratik71@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/29fe73e33415cf2b77a886eab650e23163e5571f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/29fe73e33415cf2b77a886eab650e23163e5571f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
