)]}'
{"id":"openstack%2Fblazar-tempest-plugin~739314","triplet_id":"openstack%2Fblazar-tempest-plugin~master~I482ac98bc56f0e3cfb8b767f47649da11ed1afab","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I482ac98bc56f0e3cfb8b767f47649da11ed1afab","subject":"migrate testing to ubuntu focal","status":"ABANDONED","created":"2020-07-04 17:47:42.000000000","updated":"2020-07-09 20:16:01.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"43aface9b66fe8121101b629e46d36910f4d4a1a","_number":739314,"virtual_id_number":739314,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-09 18:42:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-05 13:01:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-09 18:42:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87320e2b95ccb04357df75affb6866fbde997f03","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-04 17:47:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8758b2142926f00f9f4a41e08a8d2f9828eb05a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 18:36:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74ba5c2a91c844d88ee844b3a563e653 : FAILURE in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3694f2239d4135ae945f35d74c8bc5 : SUCCESS in 6m 02s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/2729143b84244241a96e0e1fdff2c704 : SUCCESS in 36m 56s\n- blazar-devstack-dsvm-ussuri https://zuul.opendev.org/t/openstack/build/9cfb04a42b424fb0bbc04231e8663515 : SUCCESS in 36m 54s\n- blazar-devstack-dsvm-train https://zuul.opendev.org/t/openstack/build/a217f1571a3b4c458e9ed4a2cc980e25 : SUCCESS in 40m 28s\n- blazar-devstack-dsvm-stein https://zuul.opendev.org/t/openstack/build/1a8b64a8a93a46ef88d8381eead26202 : SUCCESS in 40m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dc3e6299db9b8e918dabf588b70df42e347c44bc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-05 02:03:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50f239f5368d11e462d9710a264608ad3df4ffca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 02:51:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ff791f4b2744b0b763b76b60dfa109 : SUCCESS in 6m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b27cb71b5034853befa5a643f592dd0 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37471e89b174615bcbeca2b6cb2b4d7 : SUCCESS in 6m 17s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/13599fd7051f4e77be7097991d288a18 : SUCCESS in 37m 39s\n- blazar-devstack-dsvm-ussuri https://zuul.opendev.org/t/openstack/build/0650c0b662574cea85e757a6df7b981e : SUCCESS in 40m 33s\n- blazar-devstack-dsvm-train https://zuul.opendev.org/t/openstack/build/e32306e1e30c4ad7b505d57536dc6e13 : SUCCESS in 41m 17s\n- blazar-devstack-dsvm-stein https://zuul.opendev.org/t/openstack/build/3c0e4259088b450aba5d138537454be2 : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d022ac225abe1959baa68520a121f8009c59536a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-05 13:01:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7ad9e1d8fee472c0a42a6d054cb2e7cef1eced8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-05 18:41:48.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7d9f0aa40dfafdb595cade101ceebaf9f88f37ea","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-05 18:42:29.000000000","message":"Patch Set 3: Workflow-1\n\nlet me release hacking 3.2.0 before we merge this.","accounts_in_message":[],"_revision_number":3},{"id":"fa66cd40258085a840078ef0fc5189771e027c91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 18:55:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9cc4329f932474391fedeb1cb5c84bf : FAILURE in 4m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/44d5da37009f41c59acad9c8a92dd2e6 : FAILURE in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38b34b5942b24959bb66fe60b89bb62b : FAILURE in 9m 26s","accounts_in_message":[],"_revision_number":3},{"id":"2dbf8e0c0d142238df01ebf43d84038789c13547","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-05 19:34:06.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"b325ffd92345f83229f26002ef4f6116dac97efc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-05 19:34:13.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"148c57fab842897809a9cbc6f5149256ce99d3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 19:45:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f08cacaf114ea0802f925d9b3b1e51 : SUCCESS in 3m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3675fb4c0b444b76b574400b2dbeeb2d : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e507c013be4ac1b10b0b64f5bff5b2 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"82279b64b16101328499821a15142a7fac676c83","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-09 18:32:05.000000000","message":"Patch Set 4: -Workflow\n\nthis is ready to go","accounts_in_message":[],"_revision_number":4},{"id":"3a0e01754389bb8016e7b117cc4c79fe16218727","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-09 18:32:24.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"c19b14de7e192530c7f027c5c80e5e37c5a73465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 18:42:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03f87e7947c446b2ae8a64301e8cb586 : SUCCESS in 4m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ebcf1c98ba8b4b42a0aa73865fdd6cb5 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a340cab02940b4ad4a7981440472b9 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":5},{"id":"48a5b51ee2faeb9e833a84e05721d42364ae3715","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-09 20:16:01.000000000","message":"Abandoned\n\ndoing in other patch due to wrong commit-id here https://review.opendev.org/#/c/740357/","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"539e174584c5777c1fdd6c06542fdd399c5959e4","revisions":{"2facb763bcf43c03c9fcc8bf34a56a0779ac5bc6":{"kind":"REWORK","_number":1,"created":"2020-07-04 17:47:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/739314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/14/739314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/1"}}},"commit":{"parents":[{"commit":"9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:46:28.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:47:41.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional tests jobs\n- docs jobs\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: I482ac98bc56f0e3cfb8b767f47649da11ed1afab\nStory: #2007865\nTask: #40178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2facb763bcf43c03c9fcc8bf34a56a0779ac5bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2facb763bcf43c03c9fcc8bf34a56a0779ac5bc6"}]},"branch":"refs/heads/master"},"ed9249354bedfe87ff2f4d5f6b38bce73923a342":{"kind":"REWORK","_number":2,"created":"2020-07-05 02:03:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/739314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/14/739314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/2"}}},"commit":{"parents":[{"commit":"9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:46:28.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-05 02:02:45.000000000","tz":-300},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1886296\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nStory: #2007865\nTask: #40178\n\nCloses-Bug: #1886296\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: I482ac98bc56f0e3cfb8b767f47649da11ed1afab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/ed9249354bedfe87ff2f4d5f6b38bce73923a342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/ed9249354bedfe87ff2f4d5f6b38bce73923a342"}]},"branch":"refs/heads/master"},"b438c71afb380657a23e21233dac3ffc6c87c7e2":{"kind":"REWORK","_number":3,"created":"2020-07-05 18:41:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/739314/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/14/739314/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/3"}}},"commit":{"parents":[{"commit":"9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:46:28.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-05 18:41:47.000000000","tz":0},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1886296\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nStory: #2007865\nTask: #40178\n\nCloses-Bug: #1886296\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: I482ac98bc56f0e3cfb8b767f47649da11ed1afab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/b438c71afb380657a23e21233dac3ffc6c87c7e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/b438c71afb380657a23e21233dac3ffc6c87c7e2"}]},"branch":"refs/heads/master"},"7d52909a8ef32017f95602c5fa462db5a5b5743a":{"kind":"REWORK","_number":4,"created":"2020-07-05 19:34:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/739314/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/14/739314/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/4"}}},"commit":{"parents":[{"commit":"9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:46:28.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-05 19:34:04.000000000","tz":0},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1886296\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nStory: #2007865\nTask: #40178\n\nCloses-Bug: #1886296\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: I482ac98bc56f0e3cfb8b767f47649da11ed1afab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/7d52909a8ef32017f95602c5fa462db5a5b5743a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/7d52909a8ef32017f95602c5fa462db5a5b5743a"}]},"branch":"refs/heads/master"},"539e174584c5777c1fdd6c06542fdd399c5959e4":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-07-09 18:32:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/739314/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/14/739314/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/14/739314/5"}}},"commit":{"parents":[{"commit":"9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9468fcf2dc10aa113161dd1fc0ac8cf1db6e640f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:46:28.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-09 18:32:21.000000000","tz":0},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1886296\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nStory: #2007865\nTask: #40178\n\nCloses-Bug: #1886296\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: I482ac98bc56f0e3cfb8b767f47649da11ed1afab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/539e174584c5777c1fdd6c06542fdd399c5959e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/539e174584c5777c1fdd6c06542fdd399c5959e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
