)]}'
{"id":"openstack%2Fblazar-tempest-plugin~788848","triplet_id":"openstack%2Fblazar-tempest-plugin~master~Ide464d5b64494050ff52f58832ed24bacecb83e9","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"wallaby-stable-jobs","hashtags":[],"change_id":"Ide464d5b64494050ff52f58832ed24bacecb83e9","subject":"Add victoria/wallaby stable branch jobs on master gate","status":"MERGED","created":"2021-04-29 23:53:39.000000000","updated":"2021-05-04 21:41:38.000000000","submitted":"2021-05-04 21:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788848-wallaby-stable-jobs","meta_rev_id":"951c717cfbafd5bd1dcb6a8e94b9aa9bae5d2b8a","_number":788848,"virtual_id_number":788848,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-04 21:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-04 21:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-04 21:03:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 00:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 21:03:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"99f029ddf36dce50c3a3f46a4e301dbac714f578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-04-29 23:53:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab6ef933a46337051a5fa7839ca306f677e737fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 00:42:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f27c3a0b8a9344c9b6ba6a6267debfcf : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f144415e5364d0ea1697bf291706e08 : SUCCESS in 5m 18s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/271ecd09e5cf4e2aaf3270cb47ca4a72 : SUCCESS in 40m 30s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/9da37f7c7ac2470abd2a515fcf00c52b : SUCCESS in 38m 52s\n- blazar-tempest-plugin-wallaby https://zuul.opendev.org/t/openstack/build/c4507a04b01f4049b789df89d3ad81d8 : SUCCESS in 42m 43s\n- blazar-tempest-plugin-victoria https://zuul.opendev.org/t/openstack/build/20c534259c37479696affc493bd0c3e5 : SUCCESS in 43m 59s\n- blazar-tempest-plugin-ussuri https://zuul.opendev.org/t/openstack/build/2dc238569fef4b07b86174f1a2d75f1f : SUCCESS in 42m 46s\n- blazar-tempest-plugin-train https://zuul.opendev.org/t/openstack/build/7af23f821ad54728a907ae823fd895b5 : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"77b974faee9f6b2872e5aff92196a87d2349bef6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-04 21:03:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c081677c837c92c7985752a531943f941e625b26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 21:03:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7de897daf2bd80ee9ac0426b979e7f0a810dbf13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 21:39:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22ab8e26d7d04024a2c927bfd1e0befe : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f486ca06e5e64e58a7d35b4efa895fcc : SUCCESS in 6m 01s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/a4e93924d0fc492eb5cefae8e1a41738 : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7ca20516e5c4734a21d9ad4f51c113e4d405fb2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 21:39:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"951c717cfbafd5bd1dcb6a8e94b9aa9bae5d2b8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 21:41:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab360cfc38814e6285befaaf3f070d5f : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44403352915f47f4264d9c8825123b6ebbabc138","revisions":{"44403352915f47f4264d9c8825123b6ebbabc138":{"kind":"REWORK","_number":1,"created":"2021-04-29 23:53:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/48/788848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/48/788848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/48/788848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/48/788848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/48/788848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/48/788848/1"}}},"commit":{"parents":[{"commit":"93422b61f443cf75e7ee7f01cc14cee7d41de594","subject":"Merge \"Change dashes to underscore in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/93422b61f443cf75e7ee7f01cc14cee7d41de594"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:53:21.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:53:38.000000000","tz":-300},"subject":"Add victoria/wallaby stable branch jobs on master gate","message":"Add victoria/wallaby stable branch jobs on master gate\n\nWe have stable/victoria and stable/wallaby released so we\nshould add their job on master gate to keep branchless\ntempest plugins compatible to those branch.\n\nThis also removes the stable/stein job as that is in EM\nstate now.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: Ide464d5b64494050ff52f58832ed24bacecb83e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/44403352915f47f4264d9c8825123b6ebbabc138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/44403352915f47f4264d9c8825123b6ebbabc138"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
