)]}'
{"id":"openstack%2Fblazar-tempest-plugin~825415","triplet_id":"openstack%2Fblazar-tempest-plugin~master~I69083849d687ddbc3f922bbf6dc42b761e4cee51","project":"openstack/blazar-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-01-27 16:09:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I69083849d687ddbc3f922bbf6dc42b761e4cee51","subject":"Add stable/xena jobs on master gate","status":"ABANDONED","created":"2022-01-20 01:25:07.000000000","updated":"2022-01-27 16:09:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84a3734a4e98623519f38868ad7de9a323bd69cd","_number":825415,"virtual_id_number":825415,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-20 02:25:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-20 08:13:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-01-27 16:06:56.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"}],"messages":[{"id":"c33f1ea1a29fbe1d1b48fe7f69bde42424893fa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-01-20 01:25:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54b858f87059a3e7c0afaaf4506fa958c6c14f13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 02:25:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8a744746b24bbcba084780d907c04b : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7766d0eb3f74a26b727307f7db2f725 : SUCCESS in 7m 22s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b538ebd38f4a49be8c5bf5baa146db01 : SUCCESS in 45m 23s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/5f1420ebfc4f433ea70c980c23c0dc6c : SUCCESS in 45m 20s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/a4df05454afc4b99a73cb78e51389b15 : SUCCESS in 44m 10s\n- blazar-tempest-plugin-wallaby https://zuul.opendev.org/t/openstack/build/fb0926c4b3a345f699616bb2a0784ff2 : SUCCESS in 48m 04s\n- blazar-tempest-plugin-victoria https://zuul.opendev.org/t/openstack/build/d1df416ac27f42e19cf60c547c302bce : SUCCESS in 46m 58s\n- blazar-tempest-plugin-ussuri https://zuul.opendev.org/t/openstack/build/accbed5e01f548648580838c664ec494 : SUCCESS in 45m 46s\n- blazar-tempest-plugin-train https://zuul.opendev.org/t/openstack/build/fa59c2d41a024a10adde5a3de9498a4c : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7961daa764403634e7a1d9f7cec86dbb98e88387","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-20 08:13:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8208e580338cc53c4e3d28a5263c91fe76a0f941","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2022-01-27 16:06:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0dd33f16a8c404aeb0b908be9e6d39a4d0641556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-27 16:09:17.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f54bf43c0b8cd09d864f5a23b0a4f3ae4f260f3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-27 16:09:39.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"84a3734a4e98623519f38868ad7de9a323bd69cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-27 16:09:51.000000000","message":"Abandoned\n\nDuplicate of https://review.opendev.org/c/openstack/blazar-tempest-plugin/+/825195","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"503503bb468419acc088decdb13c145b4afa4072","revisions":{"d5ffc4ad166a9b8544a419c0019b8614ddf2932b":{"kind":"REWORK","_number":1,"created":"2022-01-20 01:25:07.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/15/825415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/15/825415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/1"}}},"commit":{"parents":[{"commit":"362c6c520902baadf54d91ca1e0110ccc98e230d","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/362c6c520902baadf54d91ca1e0110ccc98e230d"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-01-20 01:23:23.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-01-20 01:25:00.000000000","tz":480},"subject":"Add stable/xena jobs on master gate","message":"Add stable/xena jobs on master gate\n\nWe have stable/xena released so we should add\ntheir job on master gate to keep branchless tempest\nplugins compatible to stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I69083849d687ddbc3f922bbf6dc42b761e4cee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/d5ffc4ad166a9b8544a419c0019b8614ddf2932b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/d5ffc4ad166a9b8544a419c0019b8614ddf2932b"}]},"branch":"refs/heads/master"},"138a98552ce312458d152dfc440e6fa25eca74f8":{"kind":"NO_CHANGE","_number":2,"created":"2022-01-27 16:09:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/825415/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/15/825415/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/2"}}},"commit":{"parents":[{"commit":"362c6c520902baadf54d91ca1e0110ccc98e230d","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/362c6c520902baadf54d91ca1e0110ccc98e230d"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-01-20 01:23:23.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-01-27 16:09:02.000000000","tz":60},"subject":"Add stable/xena jobs on master gate","message":"Add stable/xena jobs on master gate\n\nWe have stable/xena released so we should add\ntheir job on master gate to keep branchless tempest\nplugins compatible to stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I69083849d687ddbc3f922bbf6dc42b761e4cee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/138a98552ce312458d152dfc440e6fa25eca74f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/138a98552ce312458d152dfc440e6fa25eca74f8"}]},"branch":"refs/heads/master"},"503503bb468419acc088decdb13c145b4afa4072":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-01-27 16:09:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/825415/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/15/825415/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/15/825415/3"}}},"commit":{"parents":[{"commit":"362c6c520902baadf54d91ca1e0110ccc98e230d","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/362c6c520902baadf54d91ca1e0110ccc98e230d"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-01-20 01:23:23.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-01-27 16:09:30.000000000","tz":60},"subject":"Add stable/xena jobs on master gate","message":"Add stable/xena jobs on master gate\n\nWe have stable/xena released so we should add their job on master gate\nto keep branchless tempest plugins compatible to stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I69083849d687ddbc3f922bbf6dc42b761e4cee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/503503bb468419acc088decdb13c145b4afa4072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/503503bb468419acc088decdb13c145b4afa4072"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
