)]}'
{"id":"openstack%2Fblazar-tempest-plugin~834058","triplet_id":"openstack%2Fblazar-tempest-plugin~master~Ie8dbd0adafb9450d84ebd1e50d17c7f14a5db6bf","project":"openstack/blazar-tempest-plugin","branch":"master","hashtags":["blazar-tempest-scenario-manager-cleanup-base"],"change_id":"Ie8dbd0adafb9450d84ebd1e50d17c7f14a5db6bf","subject":"Remove setup_clients() api","status":"NEW","created":"2022-03-16 18:18:46.000000000","updated":"2022-04-04 10:55:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"63b73abc5924a07a82bc14c19a006eb4b0b7ba47","_number":834058,"virtual_id_number":834058,"owner":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-23 09:35:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-16 19:06:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59fc9152aa5d07cbda758207508ab52861c7234b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-16 18:18:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"843a247f0e3cbf12c01bf69d059fbdb57ac49d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 19:06:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee7e431845241bbb7cd78e96bf76b22 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c15a945699474b9bacbbdcc721cb05e6 : SUCCESS in 7m 29s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b4ba0c287eba414e8f2a97bc3196370b : SUCCESS in 35m 43s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/dbcdcba5e3f14a37ad5a82de077e8483 : SUCCESS in 40m 12s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/eb2043e7c8b14be48b6b372b43b01a6d : SUCCESS in 36m 18s\n- blazar-tempest-plugin-wallaby https://zuul.opendev.org/t/openstack/build/660accbd3c734f5eb657a43a78a18382 : SUCCESS in 45m 16s\n- blazar-tempest-plugin-victoria https://zuul.opendev.org/t/openstack/build/e496f34996ac447c8e2aadb597719fd6 : SUCCESS in 40m 51s\n- blazar-tempest-plugin-ussuri https://zuul.opendev.org/t/openstack/build/178c861279b442b08d24c2457678a2df : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8c2e8f8a485971a20aea7ef477485236ae03f6df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-23 08:40:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f984a552ad51f4b6b2896b78079925fbf4606936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 09:35:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e092f44900ed4a22903d202a68a4a26c : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea456618f3443d8854a34edd3014904 : SUCCESS in 8m 55s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/d99ce5c994f84ddb9db76ebad5510043 : SUCCESS in 43m 03s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/87caa4a6a0d342c493626b86e15e88ad : SUCCESS in 37m 39s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/edb4929fe3604d82979718bb95692758 : SUCCESS in 45m 01s\n- blazar-tempest-plugin-wallaby https://zuul.opendev.org/t/openstack/build/7b7b9664e1d84332942fb1b2f25590ae : SUCCESS in 46m 56s\n- blazar-tempest-plugin-victoria https://zuul.opendev.org/t/openstack/build/050303ab4acb4f7a97c752dbd2e16095 : SUCCESS in 47m 51s\n- blazar-tempest-plugin-ussuri https://zuul.opendev.org/t/openstack/build/019de6e75c8e49e394ef149f604ee35e : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":2},{"id":"01ba8f83154473a1c70462a014f15c9eb25e7edf","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-04-04 10:55:40.000000000","message":"Hashtag added: blazar-tempest-scenario-manager-cleanup-base","accounts_in_message":[],"_revision_number":2},{"id":"63b73abc5924a07a82bc14c19a006eb4b0b7ba47","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-04-04 10:55:47.000000000","message":"Topic tempest-scenario-manager-cleanup removed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac6c1b458489a440ec10dd8ab9bfc88f18c54555","revisions":{"20a07dd76024ad6568d98ff745ceaf06f9f2c894":{"kind":"REWORK","_number":1,"created":"2022-03-16 18:18:46.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/58/834058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/58/834058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/834058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/834058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/834058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/834058/1"}}},"commit":{"parents":[{"commit":"65c81be68d9eb94a49e435582125bd472befa0da","subject":"Add stable/xena jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/65c81be68d9eb94a49e435582125bd472befa0da"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-16 18:14:41.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-16 18:18:41.000000000","tz":330},"subject":"Remove setup_clients() api","message":"Remove setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: Ie8dbd0adafb9450d84ebd1e50d17c7f14a5db6bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/20a07dd76024ad6568d98ff745ceaf06f9f2c894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/20a07dd76024ad6568d98ff745ceaf06f9f2c894"}]},"branch":"refs/heads/master"},"ac6c1b458489a440ec10dd8ab9bfc88f18c54555":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-23 08:40:10.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/58/834058/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/58/834058/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/834058/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/834058/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/834058/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/834058/2"}}},"commit":{"parents":[{"commit":"41347cd00f6eaeac6ca5cf468e9cda7d09175b39","subject":"Merge \"Remove create_server_snapshot method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/41347cd00f6eaeac6ca5cf468e9cda7d09175b39"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2022-03-16 18:14:41.000000000","tz":330},"committer":{"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","date":"2022-03-23 08:40:10.000000000","tz":0},"subject":"Remove setup_clients() api","message":"Remove setup_clients() api\n\nIn tempest 27.0.0 release, Scenario manager in tempest\nhave been marked stable[1]. In Xena PTG, it has been\ndecided to start the consumption of stable-scenario-\nmanager within the tempest plugins[2].\n\n[1] https://docs.openstack.org/releasenotes/tempest/unreleased.html#relnotes-27-0-0\n[2] https://etherpad.opendev.org/p/qa-xena-priority#L48\n\nFollowing are the methods cleaned within this review:\n\n1. setup_clients()\n\nSigned-off-by: Soniya Vyas \u003csvyas@redhat.com\u003e\nChange-Id: Ie8dbd0adafb9450d84ebd1e50d17c7f14a5db6bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/ac6c1b458489a440ec10dd8ab9bfc88f18c54555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/ac6c1b458489a440ec10dd8ab9bfc88f18c54555"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
