)]}'
{"id":"openstack%2Fblazar-tempest-plugin~844132","triplet_id":"openstack%2Fblazar-tempest-plugin~master~I0930cdb342bd1c03a5ed405cf08efc587f244728","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"victoria-pin-tempest","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-06-08 14:12:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0930cdb342bd1c03a5ed405cf08efc587f244728","subject":"Remove stable/victoria job from master gate","status":"MERGED","created":"2022-05-31 21:08:25.000000000","updated":"2022-06-08 14:14:26.000000000","submitted":"2022-06-08 14:12:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844132-victoria-pin-tempest","meta_rev_id":"ff802f6b4c6ac7435fd8e0af63124164166dbb35","_number":844132,"virtual_id_number":844132,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-08 14:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-08 13:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-08 13:18:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-31 21:53:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-01 07:33:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"db831714a925ec8614de5b731a73a002cade31b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-31 21:08:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4314587248a7414e4c87bf87c3deddbec0263d7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 21:53:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c606a648b3ef4e8abd962cd9164666c0 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c1bfb84ff045b4abc38e1db26d2d5d : SUCCESS in 6m 04s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/5329d4f8e2294f13ba258fb8172e6f40 : FAILURE in 28m 13s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/8f27f440ef55460c9fe7ee72910d3298 : FAILURE in 29m 45s\n- blazar-tempest-plugin-yoga https://zuul.opendev.org/t/openstack/build/7b5b9c87ce894675a1583efeaf1f411e : SUCCESS in 35m 06s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/d1ce415562834b6d926504d0331f62ec : SUCCESS in 38m 01s\n- blazar-tempest-plugin-wallaby https://zuul.opendev.org/t/openstack/build/37981dbb41694116af88dede61d7a25c : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8fec2939bd9c728f60e8064c6e2985911590e6bf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-01 07:33:00.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1152214562ceefacb73b13340c608673ce29d009","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-08 09:36:35.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20211e2440b6cdf086b3a85e2f659a66e23c2516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 10:29:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a702fb421af44a96a64ec3ade7d005f3 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87d1d81c5bf4498bfe3289c49981552 : SUCCESS in 5m 38s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/f51f7b5074a742e981a86c2f67c8583e : SUCCESS in 44m 19s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/5b8c2e35b571462c8c384b7c676a3eea : SUCCESS in 45m 26s\n- blazar-tempest-plugin-yoga https://zuul.opendev.org/t/openstack/build/fb5534bbeb91400d9b9f4373d89db848 : SUCCESS in 43m 24s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/fc1e6ba74b1b4c37a6bfd7530c75fd2c : SUCCESS in 41m 09s\n- blazar-tempest-plugin-wallaby https://zuul.opendev.org/t/openstack/build/29487f3943444809b4b10ee8ed3cce51 : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4cad18a6abd6430624d5210825d60eb00e4de817","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-08 13:18:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6077e12e7b85315d82fdc5d9c9072b5c6fd22e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 13:19:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adef449998d0f5b09755b99d1da8340a713822fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 14:12:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc7cda7f326147ebb15c2ec590b6eee9 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38cecdfe462044e1ace57430abbf62ad : SUCCESS in 4m 51s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/53c6679d2c39404a97383ffc41f61c52 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":1},{"id":"638b9a764960f1555b4aa76ee7795eb116c3bf0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 14:12:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff802f6b4c6ac7435fd8e0af63124164166dbb35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 14:14:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47deaa302ed847c781c288d792df7ec0 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9436185ee83978c00125386ac9ba5e053d1095bc","revisions":{"9436185ee83978c00125386ac9ba5e053d1095bc":{"kind":"REWORK","_number":1,"created":"2022-05-31 21:08:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/32/844132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/32/844132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/32/844132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/32/844132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/32/844132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/32/844132/1"}}},"commit":{"parents":[{"commit":"e6dfe3dd4ca5ada870dc020b6a2a741ede0734ee","subject":"Update \u0027Maintained\u0027 stable branch jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/e6dfe3dd4ca5ada870dc020b6a2a741ede0734ee"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-31 21:02:08.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-31 21:02:08.000000000","tz":-300},"subject":"Remove stable/victoria job from master gate","message":"Remove stable/victoria job from master gate\n\nstable/victoria is in EM state and Tempest is pinned\nthere so we do not need to test the stable/victoria\njobs on master gate.\n\nChange-Id: I0930cdb342bd1c03a5ed405cf08efc587f244728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9436185ee83978c00125386ac9ba5e053d1095bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9436185ee83978c00125386ac9ba5e053d1095bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
