)]}'
{"id":"openstack%2Fblazar-tempest-plugin~864895","triplet_id":"openstack%2Fblazar-tempest-plugin~master~Ibd48ad420c3f4de43950c15769c7701a7eb49222","project":"openstack/blazar-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ibd48ad420c3f4de43950c15769c7701a7eb49222","subject":"Wait longer for lease deletion to complete","status":"MERGED","created":"2022-11-17 13:17:01.000000000","updated":"2022-11-17 15:50:32.000000000","submitted":"2022-11-17 15:49:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864895","meta_rev_id":"c08a0698f3bda9146c0208bc3511e6cf16baeecc","_number":864895,"virtual_id_number":864895,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 15:49:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-17 14:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-17 14:23:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-17 13:58:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff60da59e675ffcfe5cdfc9544df6056e89e1eff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-17 13:17:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"340384c672674d5bb9f712f5100dc13e4e8d1f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:58:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c1f818ddd354e2abe667a007f873cf8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff846a87b8e4ada9435092eaee2770c : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d759a99fb68641a7a7bc6a8162900e4d : SUCCESS in 6m 54s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c005cdde74084ec89baeba44480a3cba : SUCCESS in 35m 41s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/9309651674d5476a87566b015eef88e1 : SUCCESS in 32m 26s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/eead070397b5450f850106c0b1509a2e : SUCCESS in 37m 19s\n- blazar-tempest-plugin-yoga https://zuul.opendev.org/t/openstack/build/d6ffd39f94e64dd9ae7ac2c74f9d46b2 : SUCCESS in 32m 49s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/75e7c9923d3848fbb3b25918bb92a257 : SUCCESS in 40m 19s\n- blazar-tempest-plugin-wallaby https://zuul.opendev.org/t/openstack/build/1ca4712c1c9344f5a5564f3a40af635e : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":1},{"id":"67e62fa5dd990d4f5d4f3eec5664f18de45a3f32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-17 14:23:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"32b0a52b5893fd7ca809712237adc8da836bd0ed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-17 14:23:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a649bd1e5a700e640fc7ef62edc6efa68510e6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:05:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da3a8adb2c2c4c77911ecc2f61aa82f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e255628b41f4486b8f7cbfa5b51711f : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696cdaed7101409dadf626685d550874 : SUCCESS in 5m 02s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/dcb0a239ef8a4a9283ab975a6b18767c : SUCCESS in 31m 57s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/ffca2762dead42d4b265530032aafbd5 : SUCCESS in 28m 06s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/b569856850004d839071af2699297b02 : SUCCESS in 32m 24s\n- blazar-tempest-plugin-yoga https://zuul.opendev.org/t/openstack/build/253466d4bcfe448994e98836a20b9e31 : SUCCESS in 30m 59s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/832758620a1244fea1a0c27f34a3b4c3 : SUCCESS in 31m 21s\n- blazar-tempest-plugin-wallaby https://zuul.opendev.org/t/openstack/build/178033c67b3744c28e459abfec0a9935 : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ac6a751eecf740b998123fe7e3683e61bbbee119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:06:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c837ef6aa3bdfdbd020c4c64912e716beae0e33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:49:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89622720e1e34fcb89acc89e83258200\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c98037e1f4724a8e86a753040f0a8727 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce14875190548518b0c65cf745533af : SUCCESS in 6m 28s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/d8747c85f7e24c5e9fd66d988fb19d98 : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":2},{"id":"2f192d268d3b87a1476e473964f864e0a4715035","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:49:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c08a0698f3bda9146c0208bc3511e6cf16baeecc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:50:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09bcb64a49354cf7b994513f62c570e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99d4ccb45154cd1ae791c3775bc3248 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b04a4d07ea143226da3e20afc5c9ead574f3184d","revisions":{"399ea17bf29b87feccdbbb9df8a8013db12fb259":{"kind":"REWORK","_number":1,"created":"2022-11-17 13:17:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/95/864895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/95/864895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/95/864895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/95/864895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/95/864895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/95/864895/1"}}},"commit":{"parents":[{"commit":"5334331b20641e3278378e0777961de72175110f","subject":"Add stable/zed jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/5334331b20641e3278378e0777961de72175110f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-11-17 13:15:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-11-17 13:15:15.000000000","tz":60},"subject":"Wait longer for the lease deletion to be complete","message":"Wait longer for the lease deletion to be complete\n\nChange-Id: Ibd48ad420c3f4de43950c15769c7701a7eb49222\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/399ea17bf29b87feccdbbb9df8a8013db12fb259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/399ea17bf29b87feccdbbb9df8a8013db12fb259"}]},"branch":"refs/heads/master"},"b04a4d07ea143226da3e20afc5c9ead574f3184d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-17 14:23:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/95/864895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/95/864895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/95/864895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/95/864895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/95/864895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/95/864895/2"}}},"commit":{"parents":[{"commit":"5334331b20641e3278378e0777961de72175110f","subject":"Add stable/zed jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/5334331b20641e3278378e0777961de72175110f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-11-17 14:23:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-11-17 14:23:21.000000000","tz":60},"subject":"Wait longer for lease deletion to complete","message":"Wait longer for lease deletion to complete\n\nWhen testing with Ubuntu Jammy, this test is consistently failing. Wait\nlonger for Blazar to complete lease deletion, since server deletion is\nonly the first part of it.\n\nChange-Id: Ibd48ad420c3f4de43950c15769c7701a7eb49222\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/b04a4d07ea143226da3e20afc5c9ead574f3184d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/b04a4d07ea143226da3e20afc5c9ead574f3184d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
