)]}'
{"id":"openstack%2Fblazar-tempest-plugin~885680","triplet_id":"openstack%2Fblazar-tempest-plugin~master~I5b43a1a3061367a2d49b2d8dc10a06da6742bdaf","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"2023-1-stable-job","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-06-09 12:22:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b43a1a3061367a2d49b2d8dc10a06da6742bdaf","subject":"Add stable/2023.1 jobs on master gate","status":"MERGED","created":"2023-06-09 04:11:27.000000000","updated":"2023-06-09 12:24:10.000000000","submitted":"2023-06-09 12:22:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885680-2023-1-stable-job","meta_rev_id":"cfa39585189f98a4ccedda4cd1b85e0d490268fc","_number":885680,"virtual_id_number":885680,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-09 12:22:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-09 09:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-09 09:04:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-09 05:03:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-09 09:04:53.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"7fb31912cacb128a2ce9ea0da4a98b3162030611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-06-09 04:11:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be5785778f6a3b91a9c1b873ca363953ed449b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 05:03:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b27e3cb541a842e7843945e2a30065f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b22e8ac9e7432ea33a2d80e95bd6be : SUCCESS in 2m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25750b6f7eea4a0b99b388d9c42078c2 : SUCCESS in 4m 16s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/3d8d990c098747629d0d20636fcd6a32 : SUCCESS in 31m 50s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/067cc54884e04cf1884af383111695f4 : SUCCESS in 42m 43s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/d1d9626dee5c4b4e961061d7b682df7e : SUCCESS in 31m 39s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/d2d031fd7db643689c655a81747bdf41 : SUCCESS in 43m 10s\n- blazar-tempest-plugin-yoga https://zuul.opendev.org/t/openstack/build/20a25595426b486bb7d016a85fe5c72a : SUCCESS in 45m 19s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/62a5b8f29be74c57b1e1b9e3cf0f238e : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7e5399cb02d7c461578a9f5add82aed94dc55f5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-06-09 05:07:44.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"aeeb481dc62db086a9905f2040be26a9b56245b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 06:01:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fd86e6806b04b73903264eed28906b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42b9ea40a9d43968de844cb7e99836f : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b003c536ee34647916c03ab9999a2a9 : SUCCESS in 6m 22s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/232367345f0d44e1ba9965c17888d46a : SUCCESS in 45m 19s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/385c44411f614f04b0ec3d9757b0a5ff : SUCCESS in 35m 37s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/f13951108e924177a93735450e1c35ae : SUCCESS in 42m 54s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/672937d838b343a69b9c7c196e63b866 : SUCCESS in 29m 45s\n- blazar-tempest-plugin-yoga https://zuul.opendev.org/t/openstack/build/130dbce5326844cfbcf1ed1a68307791 : SUCCESS in 35m 37s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/bfcdd4c0f3594d63a7f18e17fbe20970 : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":2},{"id":"85427bcd2be403029c895af254d256e38bbdc336","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-09 09:04:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ad0415d59967b273b88f3d58563a0163a4da7998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 09:05:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cab36870f6ee2837438956546cdf1166bbc0f50f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 09:55:44.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80450dd27257483abe4a73ba0ada186d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca61372609b8476aa10a94ccdd224b22 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaaa13dfd3824322bd555cffcea57895 : SUCCESS in 4m 16s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1cc9d15070834020990d3244326f28d3 : FAILURE in 38m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a813f58bc84d205c4c8e1a9c067c310eca22b230","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-09 11:00:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81ced6f32199c9a9ad385b22f0c4b866bc640e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 11:50:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bca7daa575544c03a124252d2642e0f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6615cdd42542359e7a28de68e40d7a : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/508dba9eb5734bdf9475ee9dd5c2f5d5 : SUCCESS in 4m 41s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/59529dfda7b347b1b747bc3b841be556 : SUCCESS in 39m 28s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/52502ecf3f9b4a818363c1027da1a6a8 : SUCCESS in 35m 35s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/9d21c311c51640879bd562981364cf7d : SUCCESS in 36m 15s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/26ff0972c4684fe8beb78fd2a0a94b28 : SUCCESS in 44m 41s\n- blazar-tempest-plugin-yoga https://zuul.opendev.org/t/openstack/build/b2915ad813df4699b47823ec4fee6695 : SUCCESS in 32m 59s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/f00c2bb7fc1f40889741d01f368e0a6e : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a3982f1bf4cabfae8e555ca49a6de9ede9b70305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 11:50:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"44ee3940c6bd88e1c8dc9980f91849a8a4f9791b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 12:22:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a1c7842afe34bf2b8cf531fdc73d234\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613cf291825e4a66a8faaaf8b5f955fb : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8678336c354a54ac619d3c99dbf638 : SUCCESS in 3m 53s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/77c315621e2b470f902c7976340b3a66 : SUCCESS in 30m 18s","accounts_in_message":[],"_revision_number":2},{"id":"cce7b16ee0f4c33d4b43821fe22ec4c209690e91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 12:22:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cfa39585189f98a4ccedda4cd1b85e0d490268fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 12:24:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36806355dc074e73a047abc02b90983d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d6a3f3525ce415087a3b46df1013e3e : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2758d56872b5932925dee702abb9a57c6d1a123a","revisions":{"6117d81ff5a76869de05c7c0f9e9b4af5ac56826":{"kind":"REWORK","_number":1,"created":"2023-06-09 04:11:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/80/885680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/80/885680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/80/885680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/80/885680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/80/885680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/80/885680/1"}}},"commit":{"parents":[{"commit":"190a7c165e8bcf738e87e5d32035c6caa123f76b","subject":"Remove stable/wallaby job from master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/190a7c165e8bcf738e87e5d32035c6caa123f76b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-06-09 02:08:47.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-06-09 02:08:47.000000000","tz":-300},"subject":"Add stable/zed jobs on master gate","message":"Add stable/zed jobs on master gate\n\nAs zed is released, we should add its job on master\ngate to keep branchless tempest plugins compatible\nto stable branch.\n\nAlso, remove stable/wallaby jobs given that wallaby went\nto Extended Maintenance on 2022-11-07T10:52:23+00:00\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I5b43a1a3061367a2d49b2d8dc10a06da6742bdaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/6117d81ff5a76869de05c7c0f9e9b4af5ac56826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/6117d81ff5a76869de05c7c0f9e9b4af5ac56826"}]},"branch":"refs/heads/master"},"2758d56872b5932925dee702abb9a57c6d1a123a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-06-09 05:07:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/80/885680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/80/885680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/80/885680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/80/885680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/80/885680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/80/885680/2"}}},"commit":{"parents":[{"commit":"190a7c165e8bcf738e87e5d32035c6caa123f76b","subject":"Remove stable/wallaby job from master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/190a7c165e8bcf738e87e5d32035c6caa123f76b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-06-09 02:08:47.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-06-09 05:07:43.000000000","tz":0},"subject":"Add stable/2023.1 jobs on master gate","message":"Add stable/2023.1 jobs on master gate\n\nAs 2023.1 is released, we should add its job on master\ngate to keep branchless tempest plugins compatible\nto stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I5b43a1a3061367a2d49b2d8dc10a06da6742bdaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2758d56872b5932925dee702abb9a57c6d1a123a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2758d56872b5932925dee702abb9a57c6d1a123a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
