)]}'
{"id":"openstack%2Fblazar-tempest-plugin~908416","triplet_id":"openstack%2Fblazar-tempest-plugin~master~Id087aeab1c3e5baf540f99ed843e4c3cfc763006","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-14 06:58:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id087aeab1c3e5baf540f99ed843e4c3cfc763006","subject":"Remove six","status":"MERGED","created":"2024-02-08 12:49:37.000000000","updated":"2024-02-14 06:59:52.000000000","submitted":"2024-02-14 06:58:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908416-remove-six","meta_rev_id":"b05bc0bbd2a244eecb17f5c4ae63424763999ca2","_number":908416,"virtual_id_number":908416,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-14 06:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-14 06:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-14 06:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-08 13:06:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-02-08 13:45:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0a4a66f9dfec92bc261efe8ac532312e1521ad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-08 12:49:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6443228db84001cd48f20b0b4d1f810549ec2f1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-08 13:06:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"315ab36f3a6cf50b946da2f7e2d9ce1ff258259c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 13:45:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ae9b512ec46432a820b3be09b2c2486\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f07df14d1c148e382ccfe85506bd0a1 : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cce1f5a0b9c6483aaf01b58931498aef : SUCCESS in 1m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266b5f4e57f641199d836c0e5146ed5a : SUCCESS in 4m 29s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/60a16b3a64c54370bdb83f8997f910de : SUCCESS in 41m 28s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/dde7502c5bd54754b72e5ec6f513617a : SUCCESS in 45m 01s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/f44f0cfbe8e04872bc0bd68d96e3b650 : SUCCESS in 40m 33s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/a4351df0e27d438a8557cbcd70958617 : SUCCESS in 45m 49s\n- blazar-tempest-plugin-yoga https://zuul.opendev.org/t/openstack/build/2e44d9a262e443cb98027a449c7890e2 : FAILURE in 26m 40s\n- blazar-tempest-plugin-xena https://zuul.opendev.org/t/openstack/build/c1288f51dac440dc8493634efab65b66 : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b8c80c2d05269f1c313c69006d455b777353ef9b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-08 17:13:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28c53e6d808998b8eb16d9e7fb5d4b3416bfa3b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-13 23:20:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"cb8bc813132d9e3291c172f6b53e4851f6eb124a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 00:10:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4a4f55d78954fe6a51320624627de4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06ec32245f1e4fc8ab9bdbd0aa2dcbec : SUCCESS in 2m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/322001084f664d2485ee84a9c4e48f21 : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde05b33a0894356a4b0191746350a7b : SUCCESS in 3m 35s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/33076c5234154fdb9d0dfbc772c429a4 : SUCCESS in 43m 49s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/f3fbfa568a9344d89f89474a09f82b6e : SUCCESS in 44m 52s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/b09db40d785b4926bb5829047723348e : SUCCESS in 35m 43s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/d0b75fd1913f43d29c819df405eda09f : SUCCESS in 36m 19s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/47dfc49ac9d647e281299dadba924bc0 : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c24f811dad5b604831efff3b6dbe910206b8483e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-14 06:25:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e55e0cdc8dcc2eb90eb25bcb38811dea64421f2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 06:25:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e32cea316032fa047a00c3c8171241dc0739409b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 06:58:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa28440b7e6d49a5a23fd8f2a1a0ff84\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d9523f597244c84bcd74c72db657912 : SUCCESS in 2m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d00ccc31a7254e768718acc2b3b8f542 : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd2137ac09de45c5a5fb57ef4543df80 : SUCCESS in 4m 41s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/abba4ca0a05047809cb1c4ba7bb61a1c : SUCCESS in 32m 44s","accounts_in_message":[],"_revision_number":2},{"id":"446c23b6b4195982767c512a640e8c1a258736a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 06:58:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b05bc0bbd2a244eecb17f5c4ae63424763999ca2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 06:59:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50f2e99f04b94eeb867b9447e756b53b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd03b26d7444bd884ec7c9a60b396a4 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c5ef0113dfcd34b40ee039b250883ca19640d108","revisions":{"174b49fd8f9abc64ef174773b6222e13598667cb":{"kind":"REWORK","_number":1,"created":"2024-02-08 12:49:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/908416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/16/908416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/16/908416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/16/908416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/16/908416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/16/908416/1"}}},"commit":{"parents":[{"commit":"9dd53ae3f9b620c9fcd577b5d10d961500204335","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9dd53ae3f9b620c9fcd577b5d10d961500204335"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-08 12:48:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-08 12:49:34.000000000","tz":540},"subject":"Remove six","message":"Remove six\n\nThis plugin no longer supports python 2 thus usage of six is no longer\nneeded.\n\nChange-Id: Id087aeab1c3e5baf540f99ed843e4c3cfc763006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/174b49fd8f9abc64ef174773b6222e13598667cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/174b49fd8f9abc64ef174773b6222e13598667cb"}]},"branch":"refs/heads/master"},"c5ef0113dfcd34b40ee039b250883ca19640d108":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-13 23:20:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/908416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/16/908416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/16/908416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/16/908416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/16/908416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/16/908416/2"}}},"commit":{"parents":[{"commit":"63064a5b47bebb8b595d9348dba17e16ec3316ca","subject":"Update stable branch testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/63064a5b47bebb8b595d9348dba17e16ec3316ca"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-08 12:48:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-13 23:20:04.000000000","tz":0},"subject":"Remove six","message":"Remove six\n\nThis plugin no longer supports python 2 thus usage of six is no longer\nneeded.\n\nChange-Id: Id087aeab1c3e5baf540f99ed843e4c3cfc763006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/c5ef0113dfcd34b40ee039b250883ca19640d108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/c5ef0113dfcd34b40ee039b250883ca19640d108"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
