)]}'
{"id":"openstack%2Fblazar-tempest-plugin~917104","triplet_id":"openstack%2Fblazar-tempest-plugin~master~Iffc0199dcdb88504ff75179c444c3706968b1ba0","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"2024-1-stable-job","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-05-28 17:40:13.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-05-28 16:47:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-05-28 17:40:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iffc0199dcdb88504ff75179c444c3706968b1ba0","subject":"Add stable/2024.1 jobs on master gate","status":"MERGED","created":"2024-04-26 00:33:04.000000000","updated":"2024-05-28 17:42:42.000000000","submitted":"2024-05-28 17:40:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917104-2024-1-stable-job","meta_rev_id":"13681f5cd32f4998c6ae131877459564b35ffadf","_number":917104,"virtual_id_number":917104,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 17:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-28 16:47:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2024-05-28 16:00:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-28 16:47:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-26 01:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-26 07:56:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-05-02 04:13:59.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"477564903a6e7ce468e34b80a23d8f7ee66bc52d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-26 00:33:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eff272ae58166f375b53b6dd98c90d008a5cd166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 01:31:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d3f29ea22a41149964f7db76383c98\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd78690e1c0473e83107826b4e64096 : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfea7674b86e405facc6fa4e34bd634f : SUCCESS in 4m 06s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e0eac3edbd3f46489023059c8debf555 : SUCCESS in 51m 07s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/1b21ad2abfa249359e8de25009f5d506 : SUCCESS in 33m 32s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/9ee85376a66c43628a5191fd643a758b : SUCCESS in 39m 36s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/a36ac519ca284c91b8b6ba9b38e55ffd : SUCCESS in 31m 22s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/8e4c480fe0964a81ae442b06fc53e04d : SUCCESS in 32m 33s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/015bee003f6a40f08d363e3fb02e81f8 : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b6b33c04f4aea47fe7154128004c348c00b35c29","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-26 07:56:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63b8af800a0b873fdde1515d14c7a39e1933d0b2","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-05-16 08:04:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8aead1366181ed75d0703b68269d3df5c3cfbe59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 08:04:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7ee1ee614a191ebc383faa972d85f0f463ef373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 08:37:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e21c7fe2a314cf89f0f3b30d20ef2ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a2ce8b796043fca89aafbc15bebf2b : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc7335c0c844176be90717bb1fbb230 : SUCCESS in 4m 24s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/cb414e53ea0c453b9db6127e5b405df9 : FAILURE in 27m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d3258dc37d67ca1c73f44bbbf6585192cd9d82bb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-16 15:19:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"864da366ec23466aae3b6d2ffc391b8d2fd859f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 16:10:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a86ebaf1dd84024972a36bcd7c41cbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d31b601d96af485086b4fc2049c2fa0c : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26e7eeb112144a98d4e3defce045232 : SUCCESS in 3m 34s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/cb0d08804d13409f9b7c26947e24a6d7 : FAILURE in 29m 40s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/4cb11d812bec4d0b9ddc9b96703c509c : FAILURE in 29m 14s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/ca3c4e130e1343189aace3d5acdfdcd7 : FAILURE in 41m 45s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/74a69c6161954454a2ec3ca960b3983c : SUCCESS in 43m 56s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/62c5ff13524f4a7f90d97c9814241be1 : SUCCESS in 45m 09s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/f7cf5e41e5ae488b8a993d19f6636a2a : FAILURE in 6m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cf5bcd31549cbc530a6d534aea016a4df13340aa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-28 07:03:10.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d63077e7f8d781519e6d7197ce20c04be9c27f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 08:00:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42ad90910ac04bc9979ebaf54cf0ea0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a309d622a147b09f94bbd4f9918a18 : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ed207205df478e965928a746f077d4 : SUCCESS in 4m 02s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/2736f2b277c04b3fad1b2b67326876a4 : SUCCESS in 32m 26s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/6eb9ef9879fd451f82429a1f76706448 : SUCCESS in 34m 06s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/e3042ec05415445bbc1ef46ed81333f2 : SUCCESS in 32m 10s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/fb3a5bb385e14aa9bdaebc15a68ce772 : SUCCESS in 32m 25s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/151137d5bdde40c8ac85dee7c01c8ef2 : SUCCESS in 33m 32s\n- blazar-tempest-plugin-zed https://zuul.opendev.org/t/openstack/build/2753bd0f369b478d87814789646425b1 : FAILURE in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"30b748451c1b271b5df8eb26c95cd0ae9d781bbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-28 16:00:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":2},{"id":"3752697337186ae97fab96a5c2bab253ee22bc3d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-28 16:47:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dca66a9628fc1522275a9ffa86214d755d6a6e53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:08:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e7c18101ddd4ebcb9446d33e265ef68\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4942a3d100342bdb8cbc4254f67d7ae : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cbcc66a6a8c4c9a9b9aa6360e5347b9 : SUCCESS in 3m 56s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/5f91757074ce473cbfe66491599f48d2 : SUCCESS in 51m 59s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/6954cf4840814e9b8a77e1f79d5bd5c8 : SUCCESS in 41m 59s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/66ab71af0e964314bf1be5aa2a7d3006 : SUCCESS in 40m 37s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/f7d4140c2bc940f3961383ac8cf39853 : SUCCESS in 30m 46s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/d3fdd19fe34a424690f521180257dd35 : SUCCESS in 36m 44s","accounts_in_message":[],"_revision_number":2},{"id":"3e4ed6d63c733e475a8909e2581eb246fd662663","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:09:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"249e8e4874d6eb7c33c549ab01670052ec7b8196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:40:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4567985ec8704eca80453547463f46b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5fdf744ff9d4c90bbfb8637bb1c996e : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d055c8f6a84efe8b1b8866b51b69bb : SUCCESS in 3m 50s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/21aba660deb048b29056cfb6894b14b4 : SUCCESS in 30m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b53a8958ddf39d42bbdea885dead476b4e93f6ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:40:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"13681f5cd32f4998c6ae131877459564b35ffadf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:42:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43abf621debb430282d61383278f3b56\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b28fa625fa4198a2f89bbb9181c0b1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12c9fbd583a25d07a0f4670e6d827c6ac57a1468","revisions":{"f301a716be39d318665f0edb697a011df8056790":{"kind":"REWORK","_number":1,"created":"2024-04-26 00:33:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/04/917104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/04/917104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/04/917104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/04/917104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/04/917104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/04/917104/1"}}},"commit":{"parents":[{"commit":"c5ef0113dfcd34b40ee039b250883ca19640d108","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/c5ef0113dfcd34b40ee039b250883ca19640d108"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-26 00:30:31.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-26 00:33:02.000000000","tz":-420},"subject":"Add stable/2024.1 jobs on master gate","message":"Add stable/2024.1 jobs on master gate\n\nAs 2024.1 is released, we should add its job on master\ngate to keep branchless tempest plugins compatible to stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: Iffc0199dcdb88504ff75179c444c3706968b1ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/f301a716be39d318665f0edb697a011df8056790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/f301a716be39d318665f0edb697a011df8056790"}]},"branch":"refs/heads/master"},"12c9fbd583a25d07a0f4670e6d827c6ac57a1468":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-28 16:00:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/917104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/04/917104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/04/917104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/04/917104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/04/917104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/04/917104/2"}}},"commit":{"parents":[{"commit":"142cd421ee71a018b2aca62934c2d1900e42fff2","subject":"Remove test for unmaintained zed branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/142cd421ee71a018b2aca62934c2d1900e42fff2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-26 00:30:31.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-05-28 16:00:51.000000000","tz":0},"subject":"Add stable/2024.1 jobs on master gate","message":"Add stable/2024.1 jobs on master gate\n\nAs 2024.1 is released, we should add its job on master\ngate to keep branchless tempest plugins compatible to stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: Iffc0199dcdb88504ff75179c444c3706968b1ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
