)]}'
{"id":"openstack%2Fblazar-tempest-plugin~927398","triplet_id":"openstack%2Fblazar-tempest-plugin~master~I54932bb3d74411e324941d295d0241ad3fbae6ac","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"flavor_plugin","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-09-03 19:24:18.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-08-29 10:48:48.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I54932bb3d74411e324941d295d0241ad3fbae6ac","subject":"Add scenario for flavor-based instance reservation","status":"MERGED","created":"2024-08-28 15:37:28.000000000","updated":"2024-09-03 20:20:41.000000000","submitted":"2024-09-03 20:18:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927398-flavor_plugin","meta_rev_id":"65bcdfc172d6ac7b1a4dd2e3619525ba5b748ba7","_number":927398,"virtual_id_number":927398,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 20:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2024-09-03 13:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":2,"date":"2024-08-29 10:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-03 19:24:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-28 17:01:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-29 08:06:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"},{"updated":"2024-08-29 08:06:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"28ab4ce9013f6403851f6ba0fdab0cd32a0e9299","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-28 15:37:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"390a9dc781ed6a8ed52fcad71afda1ad9bb75321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-28 15:46:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"59ed3ad406ad3fbee1b09c4c5ccd10efd7c9b7e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-28 15:48:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7111d8b29f4885f55d4350a0df4f46333c5a4579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-28 16:22:21.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"bd914d8ae5303ada27d5834abe69f43cd56b0da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:01:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c7d1f18f9be49f884f4c7d4edd45ae0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01f99dad96f4460a46775ad994356df : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e3999340d449d481f4048f68b9f7c2 : SUCCESS in 4m 32s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/0512b3be091c4e17b1427e978e55e492 : SUCCESS in 32m 00s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/556e6e098ea14b85b143f97cfc59268a : SUCCESS in 33m 38s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/90010a4838da4da98537512f7cc3e210 : FAILURE in 22m 49s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/662a23c5c4ae493aa7c57b467ab26fac : FAILURE in 21m 00s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/ae1ad7edbb2541589bf245f7a429c739 : FAILURE in 25m 36s","accounts_in_message":[],"_revision_number":4},{"id":"18261c304841bbc171a0bb99f5a51fd211025d70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-28 20:35:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b9a7e7831b69c524830c57d14dedde53a76a7c40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-28 21:07:41.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"37af31e808acd88e4700e805cf330309ff4d33da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 21:51:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/031e1073c3494f1d9af1c7a9ec7fe469\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a0de221bf34399a1abd7f323c41287 : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c8861e35d494a38bbc31cb7c5b59c23 : SUCCESS in 3m 50s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/266ef505edbd42bf90d0be4a885217b9 : SUCCESS in 34m 18s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/3221b7dc377844c68cdda994d4282368 : SUCCESS in 38m 29s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/c75bd7bdca0a44678c397999a409fb95 : SUCCESS in 42m 34s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/2e9070ceac19472a9404bc4369a966e0 : SUCCESS in 39m 37s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/a072a7525bed4c35b60e2537d38b135f : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":6},{"id":"698c88315434b2427d46a4bbabef735225ab21ce","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-29 08:55:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2e197c64eb6d18c0af26d8e681e76b77bca93912","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-29 09:40:43.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"102e82661818d57bf32caada0ecbec8d87999550","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-29 09:40:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"db465acb86b070c902c9ec217dd930676949b2d4","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-29 09:46:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f35805d0e2b880386424e527934a60101283334a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-29 09:50:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"16bab056c30372ced94ea00e74792798cc5cebdf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-29 09:51:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5aa520a39eda73fa327ffbc7f148c2baad07ad44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 10:43:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c83974ca25db483c890ada008f622a31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6006a9aa004c9cb0c2a037e81aae00 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc76e4bebb9418e84da6ec8ee49019c : SUCCESS in 3m 51s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/28b52ba8910f4d29afbdb42484a45ece : SUCCESS in 33m 44s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/7fdc78d5146f4ae1a782e67bc36112e3 : SUCCESS in 32m 11s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/eac8984310f346abae2d052b08082424 : SUCCESS in 34m 37s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/524a0a3dd8df44e5b8ff0ecee42dee58 : SUCCESS in 31m 42s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/fe04f05e093d4c259d98c7600948e4b6 : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":8},{"id":"2a2336b6cb7687f067f84f7a13c11369056bcd0b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-29 10:48:48.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bceb66b830160f221c1387177551acabc63500fe","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2024-09-03 13:49:56.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"d55b9cb5b0aa3fc357d8f8c82348e326d2bab275","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-03 19:24:18.000000000","message":"Patch Set 8: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5da5d73dc2d0831c5484d57bd59fb16400544abd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 19:24:56.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"e1c7b73521ec00063f9f984075914dd6ae8b20cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 20:18:13.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/700b0f158e1d45958050f94b740acab8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea411948c684e71905d92b409810d78 : SUCCESS in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08af5b5a5d4343a1b0f30934a55353e8 : SUCCESS in 4m 31s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/fd97314c7eb7412293d6d69d10e53098 : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":8},{"id":"a54131faa7dd66028338484a6d88ce5ff8782b9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 20:18:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"65bcdfc172d6ac7b1a4dd2e3619525ba5b748ba7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 20:20:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/422e9aff80a543629c90e2100718d073\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87bcb810d9a45da99684fbab290ef3e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"2950909d9b871ffb7ac98039d2c6743be620383f","revisions":{"897bceae6622e7207ecfca9cf50f018374970bc7":{"kind":"REWORK","_number":1,"created":"2024-08-28 15:37:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/927398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/98/927398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/1"}}},"commit":{"parents":[{"commit":"12c9fbd583a25d07a0f4670e6d827c6ac57a1468","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 15:37:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 15:37:20.000000000","tz":120},"subject":"Add scenario for flavor-based instance reservation","message":"Add scenario for flavor-based instance reservation\n\nThis scenario is only enabled for the master branch, which will soon\nbecome stable/2024.2.\n\nChange-Id: I54932bb3d74411e324941d295d0241ad3fbae6ac\nDepends-On: https://review.opendev.org/c/openstack/blazar/+/927016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/897bceae6622e7207ecfca9cf50f018374970bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/897bceae6622e7207ecfca9cf50f018374970bc7"}]},"branch":"refs/heads/master"},"9035161c31f4e96b7ebf38a50f52ba427b7aaf9e":{"kind":"REWORK","_number":2,"created":"2024-08-28 15:46:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/927398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/98/927398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/2"}}},"commit":{"parents":[{"commit":"12c9fbd583a25d07a0f4670e6d827c6ac57a1468","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 15:46:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 15:46:31.000000000","tz":120},"subject":"Add scenario for flavor-based instance reservation","message":"Add scenario for flavor-based instance reservation\n\nThis scenario is only enabled for the master branch, which will soon\nbecome stable/2024.2.\n\nChange-Id: I54932bb3d74411e324941d295d0241ad3fbae6ac\nDepends-On: https://review.opendev.org/c/openstack/blazar/+/927016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9035161c31f4e96b7ebf38a50f52ba427b7aaf9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/9035161c31f4e96b7ebf38a50f52ba427b7aaf9e"}]},"branch":"refs/heads/master"},"a78f523ddbec6d99578fad1b4599c2fa5afcb119":{"kind":"REWORK","_number":3,"created":"2024-08-28 15:48:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/927398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/98/927398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/3"}}},"commit":{"parents":[{"commit":"12c9fbd583a25d07a0f4670e6d827c6ac57a1468","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 15:48:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 15:48:47.000000000","tz":120},"subject":"Add scenario for flavor-based instance reservation","message":"Add scenario for flavor-based instance reservation\n\nThis scenario is only enabled for the master branch, which will soon\nbecome stable/2024.2.\n\nChange-Id: I54932bb3d74411e324941d295d0241ad3fbae6ac\nDepends-On: https://review.opendev.org/c/openstack/blazar/+/927016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/a78f523ddbec6d99578fad1b4599c2fa5afcb119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/a78f523ddbec6d99578fad1b4599c2fa5afcb119"}]},"branch":"refs/heads/master"},"88c9542f743656de807bfed035c8d4b8c52d3308":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-08-28 16:22:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/927398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/98/927398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/4"}}},"commit":{"parents":[{"commit":"12c9fbd583a25d07a0f4670e6d827c6ac57a1468","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 16:22:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 16:22:12.000000000","tz":120},"subject":"Add scenario for flavor-based instance reservation","message":"Add scenario for flavor-based instance reservation\n\nThis scenario is only enabled for the master branch, which will soon\nbecome stable/2024.2.\n\nChange-Id: I54932bb3d74411e324941d295d0241ad3fbae6ac\nDepends-On: https://review.opendev.org/c/openstack/blazar/+/927413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/88c9542f743656de807bfed035c8d4b8c52d3308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/88c9542f743656de807bfed035c8d4b8c52d3308"}]},"branch":"refs/heads/master"},"41171dd8dcc339b41276044d5eaac09e4f9934ee":{"kind":"REWORK","_number":5,"created":"2024-08-28 20:35:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/927398/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/98/927398/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/5"}}},"commit":{"parents":[{"commit":"12c9fbd583a25d07a0f4670e6d827c6ac57a1468","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 20:34:54.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 20:34:54.000000000","tz":120},"subject":"Add scenario for flavor-based instance reservation","message":"Add scenario for flavor-based instance reservation\n\nThis scenario is only enabled for the master branch, which will soon\nbecome stable/2024.2.\n\nChange-Id: I54932bb3d74411e324941d295d0241ad3fbae6ac\nDepends-On: https://review.opendev.org/c/openstack/blazar/+/927413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/41171dd8dcc339b41276044d5eaac09e4f9934ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/41171dd8dcc339b41276044d5eaac09e4f9934ee"}]},"branch":"refs/heads/master"},"c8c1b80746ebb43d8beef03c6e9831cb88bf74db":{"kind":"REWORK","_number":6,"created":"2024-08-28 21:07:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/927398/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/98/927398/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/6"}}},"commit":{"parents":[{"commit":"12c9fbd583a25d07a0f4670e6d827c6ac57a1468","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 20:34:54.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-28 21:07:41.000000000","tz":0},"subject":"Add scenario for flavor-based instance reservation","message":"Add scenario for flavor-based instance reservation\n\nThis scenario is only enabled for the master branch, which will soon\nbecome stable/2024.2.\n\nChange-Id: I54932bb3d74411e324941d295d0241ad3fbae6ac\nDepends-On: https://review.opendev.org/c/openstack/blazar/+/927413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/c8c1b80746ebb43d8beef03c6e9831cb88bf74db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/c8c1b80746ebb43d8beef03c6e9831cb88bf74db"}]},"branch":"refs/heads/master"},"da5b00b08975de49dfcc178a50d857a8e13953fc":{"kind":"REWORK","_number":7,"created":"2024-08-29 09:40:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/927398/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/98/927398/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/7"}}},"commit":{"parents":[{"commit":"12c9fbd583a25d07a0f4670e6d827c6ac57a1468","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-29 09:40:38.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-29 09:40:38.000000000","tz":120},"subject":"Add scenario for flavor-based instance reservation","message":"Add scenario for flavor-based instance reservation\n\nThis scenario is only enabled for the master branch, which will soon\nbecome stable/2024.2.\n\nChange-Id: I54932bb3d74411e324941d295d0241ad3fbae6ac\nDepends-On: https://review.opendev.org/c/openstack/blazar/+/927413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/da5b00b08975de49dfcc178a50d857a8e13953fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/da5b00b08975de49dfcc178a50d857a8e13953fc"}]},"branch":"refs/heads/master"},"2950909d9b871ffb7ac98039d2c6743be620383f":{"kind":"REWORK","_number":8,"created":"2024-08-29 09:50:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/927398/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/98/927398/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/98/927398/8"}}},"commit":{"parents":[{"commit":"12c9fbd583a25d07a0f4670e6d827c6ac57a1468","subject":"Add stable/2024.1 jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/12c9fbd583a25d07a0f4670e6d827c6ac57a1468"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-29 09:50:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-08-29 09:50:31.000000000","tz":120},"subject":"Add scenario for flavor-based instance reservation","message":"Add scenario for flavor-based instance reservation\n\nThis scenario is enabled by default, but the configuration option\nCONF.resource_reservation.flavor_instance_plugin can be set to False to\nskip it.\n\nZuul configuration of stable branches are updated to skip this scenario.\n\nChange-Id: I54932bb3d74411e324941d295d0241ad3fbae6ac\nDepends-On: https://review.opendev.org/c/openstack/blazar/+/927413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2950909d9b871ffb7ac98039d2c6743be620383f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2950909d9b871ffb7ac98039d2c6743be620383f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
