)]}'
{"id":"openstack%2Fblazar-tempest-plugin~933358","triplet_id":"openstack%2Fblazar-tempest-plugin~master~I54a14f2d7ae1689d690bdef892cb721dcde88d0c","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"support-py312","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-10-29 19:42:11.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-10-28 15:44:09.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I54a14f2d7ae1689d690bdef892cb721dcde88d0c","subject":"Support py3.12 and drop py3.8","status":"MERGED","created":"2024-10-24 19:20:21.000000000","updated":"2024-10-29 19:43:10.000000000","submitted":"2024-10-29 19:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933358-support-py312","meta_rev_id":"8b343b4a8d530470da8338c0d4914d767be2c4ce","_number":933358,"virtual_id_number":933358,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-29 19:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-28 15:44:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2024-10-29 19:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2024-10-29 19:09:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 20:15:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-24 20:19:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-10-29 19:09:37.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"de6958569b6cc93a92fa7942f38d3faf12b26229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-24 19:20:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edcd5ac33cf2a692afa38dcc7962e04b3e163902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:15:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbbd4a763fe3414892f1890fc97822d4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4fe0396dd874673a63cfe4398deea62 : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2284c4571bca45abbc92f703b8642d4c : SUCCESS in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40053624c26c40fb8ee9a07553a4f1bc : SUCCESS in 5m 10s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/f6e967a2a6594f47ae17bbb52815ea59 : SUCCESS in 43m 43s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/70b3d558ef2f4a5a8d605fe1ffef4f3a : SUCCESS in 41m 49s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/378b88a2a1cd4a6db76f0bd8e521ef76 : SUCCESS in 46m 12s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/986afd474b694d2cae17147a491e6e75 : SUCCESS in 26m 21s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/4ab660118e5245aa9f3e1295b22285d6 : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2972df271b1f5efa78914e5b1dce3d9559d51d2b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-24 20:19:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7f662483d6c70092cdc115e800b1f8cb3434c41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-25 22:25:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9af14a49d46f8449f72aeb1804105f8bb03a6f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 23:14:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ace11b811764e4db61c4ad76545986d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/144c78a31bce49639c84a43d508e38e6 : SUCCESS in 1m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc8610566cae42859a7188430f1bfbb3 : SUCCESS in 2m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba812a7281044839bf7dfc28000e3205 : SUCCESS in 3m 47s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/60f714a3738944dba6f4c24f290c7db3 : SUCCESS in 40m 24s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/72a391cf8c9d44d0adc06734a4c417ef : SUCCESS in 40m 14s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/290bfe1d9601477d895e171e6c7e020f : SUCCESS in 42m 51s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/59a4aa48850a4f549e6633eb6b0e5b55 : SUCCESS in 40m 14s\n- blazar-tempest-plugin-2023-1 https://zuul.opendev.org/t/openstack/build/face08e2dcfe4794a933d83091220228 : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b4fbda13316fd306cb0d0922da20231b0aed3948","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-28 15:44:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"289458cfb0353806ef054c8b9793290088e71e11","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-10-29 19:09:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"335a22b40324e6cdb6bcf471e82f77fba46a55c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 19:10:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3c14a4ae27f57c5ca8bfb4a09c513b0eeea63d36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 19:42:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c466af9685f7494ba35d8c6a54bd88b7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2da106e78fae43949c925f0c08814a02 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2aab0e4c1142e6b0c8e2874a9bb010 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d7c891b2e6450bade80492b2137522 : SUCCESS in 5m 57s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/30d2751e713f4c44b92494590a10532b : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":2},{"id":"842980954cf1e50cdcdff066071f283b975a2ffd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 19:42:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8b343b4a8d530470da8338c0d4914d767be2c4ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 19:43:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d399a17ac642f2be8d5819eca61bec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac5451fb37241488bc5c56c8f6c0021 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b5e8f346968eb9ef547e8b2bd9bf979d2b1f054","revisions":{"dee9f1f4379842a7a4735b3a827ef2f3dec63881":{"kind":"REWORK","_number":1,"created":"2024-10-24 19:20:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/933358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/58/933358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/933358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/933358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/933358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/933358/1"}}},"commit":{"parents":[{"commit":"2950909d9b871ffb7ac98039d2c6743be620383f","subject":"Add scenario for flavor-based instance reservation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2950909d9b871ffb7ac98039d2c6743be620383f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 19:12:33.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 19:12:33.000000000","tz":-420},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nIn tempest plugin, we test from python 3.8 till\npython 3.12 so updating the same in python\nclassifier in setup.cfg\n\nChange-Id: I54a14f2d7ae1689d690bdef892cb721dcde88d0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/dee9f1f4379842a7a4735b3a827ef2f3dec63881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/dee9f1f4379842a7a4735b3a827ef2f3dec63881"}]},"branch":"refs/heads/master"},"2b5e8f346968eb9ef547e8b2bd9bf979d2b1f054":{"kind":"REWORK","_number":2,"created":"2024-10-25 22:25:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/933358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/58/933358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/933358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/933358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/933358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/58/933358/2"}}},"commit":{"parents":[{"commit":"2950909d9b871ffb7ac98039d2c6743be620383f","subject":"Add scenario for flavor-based instance reservation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2950909d9b871ffb7ac98039d2c6743be620383f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 19:12:33.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-25 22:17:02.000000000","tz":-420},"subject":"Support py3.12 and drop py3.8","message":"Support py3.12 and drop py3.8\n\nAs per the current release tested runtime, we test\nfrom python 3.9 till python 3.12 so updating the same\nin python classifier in setup.cfg\n\nAlso, dropping the python3.8 support\n- https://lists.openstack.org/archives/list/openstack-discuss@li\u003e\n\nChange-Id: I54a14f2d7ae1689d690bdef892cb721dcde88d0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2b5e8f346968eb9ef547e8b2bd9bf979d2b1f054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/2b5e8f346968eb9ef547e8b2bd9bf979d2b1f054"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
