)]}'
{"id":"openstack%2Fblazar-tempest-plugin~934277","triplet_id":"openstack%2Fblazar-tempest-plugin~master~I792d6007725d0373daa9cf42218a6f28a316560c","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"2023-1-last","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-22 18:05:18.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-11-22 17:07:53.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I792d6007725d0373daa9cf42218a6f28a316560c","subject":"Remove stable/2023.1 job from master gate","status":"MERGED","created":"2024-11-06 19:51:32.000000000","updated":"2024-11-22 18:08:26.000000000","submitted":"2024-11-22 18:05:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934277-2023-1-last","meta_rev_id":"59246a306e216314fecfe6bd4de4d51d666aa2b0","_number":934277,"virtual_id_number":934277,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-22 18:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-19 08:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2024-11-22 17:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2024-11-22 17:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-06 20:50:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-07 21:54:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-11-19 08:21:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-11-19 08:21:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"fd67c57368c19fb15d20ca184bfb2f796d469b46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-06 19:51:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54336f2b1ba64e09a12f1931e83a59d66fb985eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 20:50:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d10a55098d43368832790cb7fe5af9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/071e54417a7e44299dacd8fbf81c287e : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc67dd8fe2a40dfb1f8d0b2a6232b1d : SUCCESS in 2m 59s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/348086ecad4f4937b7aef4e02fc7a75e : SUCCESS in 29m 43s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/3e1bca01e1674ebf9dddb305a9e2b841 : SUCCESS in 41m 23s\n- blazar-tempest-plugin-2024-2 https://zuul.opendev.org/t/openstack/build/94dd1a1b45324294a6ef6406395293dc : SUCCESS in 33m 05s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/4e19a46f34ca46ed89d651a4bbeac98b : SUCCESS in 27m 11s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/6bc7997b1a1848b4ab2a8b1cc91c0b03 : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2b76ffb3df2d96b109fce2cd031776269ace3ff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-07 17:48:30.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3e0a3fe1b643b3165673d4898ba7a0f139bdae34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 18:35:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67166803d5a046ce84fd1328fc65b23a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dad211513b648caad2df22b0cbefe3f : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e41acd28b714c2396e7554ab4363891 : SUCCESS in 4m 03s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/7dd867646b764ae6866ff7b8aa2ff38b : FAILURE in 32m 49s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e98e8a68623f41708535dee5aefce26c : SUCCESS in 31m 35s\n- blazar-tempest-plugin-2024-2 https://zuul.opendev.org/t/openstack/build/8f0545713aff4d84882ea5b3d6661317 : SUCCESS in 30m 24s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/a1973a63503542c199f2191eb4f0ca99 : SUCCESS in 31m 14s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/55cc784596104abfbf0b9f4c76e57963 : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":2},{"id":"07754472db5a0584831b2a9836e1b9ac0a7119cd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-07 21:54:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28f321118377f6a97580d0189c1fbae03ba241a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 22:43:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbaacce482b94a529724e759706ef3be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35f2ec7d94854724b5b87c6b5a877149 : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e8f2600bcd4ab4a0142a98b5370be3 : SUCCESS in 2m 45s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/abcb2e9e2a59472aa222cd07f5a1f655 : SUCCESS in 42m 43s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/6eb828b46e2148829812b9c3b5c37d82 : SUCCESS in 31m 38s\n- blazar-tempest-plugin-2024-2 https://zuul.opendev.org/t/openstack/build/d30b35da8aba470993c4cd1fa96e40d4 : SUCCESS in 31m 32s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/274daa31f4d9438f8d624d5e57b39ae4 : FAILURE in 36m 15s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/4c8203c3cc174d03adb8d0161653e240 : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":2},{"id":"036aa8788452dbe3298f6a27a5a27ce19d2a89f7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-19 03:40:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47654f65fb2c22c3bea8e4a66bf48d98d0073b86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 04:22:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0a08232b3d64473a76e0cb8b7a30523\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4dfcc8a151b45a9a27892fdf3073899 : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17038a71ebbd4bffacdddd552d63ef3f : SUCCESS in 5m 05s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/56e4f1f47de848349bc5facc14b3d747 : SUCCESS in 31m 45s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/639d87491c8440b5b9d3439e820a8dd4 : SUCCESS in 37m 23s\n- blazar-tempest-plugin-2024-2 https://zuul.opendev.org/t/openstack/build/cba45cb36e72462b89728ad937843ed8 : SUCCESS in 30m 23s\n- blazar-tempest-plugin-2024-1 https://zuul.opendev.org/t/openstack/build/74b3bd02b7384c3185413cf7f6b33e6a : SUCCESS in 32m 09s\n- blazar-tempest-plugin-2023-2 https://zuul.opendev.org/t/openstack/build/3e508e4d6ee94434a222f101cc1e2269 : SUCCESS in 30m 22s","accounts_in_message":[],"_revision_number":2},{"id":"5d0c3ded7955173c3497129b10f6c57d0b110b52","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-19 08:21:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a64a86b684fa105c3cd53c4c571890b76933efc8","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-11-22 17:07:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e2ef8ec0b4bbe084add8dd61e59b319ce530b858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 17:08:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e58dcdada4f0dbcce7adfd5086802f0e30f675d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 18:05:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3ea9debc4bb4563b2829fe7625b115b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba2b3a4ccf244d4f9a47f4bd8d8d214d : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33a5339c09d4d39989f0a481057aa7f : SUCCESS in 4m 01s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/d1e7b26e18eb4c2da631c1cb006c10e5 : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":2},{"id":"32a9f6692ef54a83693d9ad73725b95d11b77142","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 18:05:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"59246a306e216314fecfe6bd4de4d51d666aa2b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 18:08:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be447644588841d6ae3ed52c8d22b196\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac6f71b90b524b3585374a85f875ef29 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6350fdfc87dae68571e3344289a88d45a07f535a","revisions":{"72d1b74d650c341b088710c7274f27069838fb88":{"kind":"REWORK","_number":1,"created":"2024-11-06 19:51:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/77/934277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/77/934277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/77/934277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/77/934277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/77/934277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/77/934277/1"}}},"commit":{"parents":[{"commit":"7020a704cdaf95371987b9c81e7f243512f34d37","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/7020a704cdaf95371987b9c81e7f243512f34d37"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-06 19:39:58.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-06 19:39:58.000000000","tz":-480},"subject":"Remove stable/2023.1 job from master gate","message":"Remove stable/2023.1 job from master gate\n\nstable/2023.1 is planned to be in unmaintained state[1]\nand with python 3.8 drop from Tempest[2], we have to pin the\nTempest for stable/2023.1 testing[3]. This means Tempest and\nits plugin master does not need to support stable/2023.1.\n\nWith that we can remove the stable/2023.1 jobs from tempest\nplugin master jobs.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@\u003e\n[2] https://review.opendev.org/c/openstack/tempest/+/933398\n[3] https://review.opendev.org/c/openstack/devstack/+/933704\n\nChange-Id: I792d6007725d0373daa9cf42218a6f28a316560c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/72d1b74d650c341b088710c7274f27069838fb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/72d1b74d650c341b088710c7274f27069838fb88"}]},"branch":"refs/heads/master"},"6350fdfc87dae68571e3344289a88d45a07f535a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-07 17:48:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/77/934277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/77/934277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/77/934277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/77/934277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/77/934277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/77/934277/2"}}},"commit":{"parents":[{"commit":"7020a704cdaf95371987b9c81e7f243512f34d37","subject":"Merge \"Support py3.12 and drop py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/7020a704cdaf95371987b9c81e7f243512f34d37"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-06 19:39:58.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-07 17:48:30.000000000","tz":0},"subject":"Remove stable/2023.1 job from master gate","message":"Remove stable/2023.1 job from master gate\n\nstable/2023.1 is planned to be in unmaintained state[1]\nand with python 3.8 drop from Tempest[2], we have to pin the\nTempest for stable/2023.1 testing[3]. This means Tempest and\nits plugin master does not need to support stable/2023.1.\n\nWith that we can remove the stable/2023.1 jobs from tempest\nplugin master jobs.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/UDQAC7SR5JAQJE5WBAG54A2MTBVBTJ44/\n[2] https://review.opendev.org/c/openstack/tempest/+/933398\n[3] https://review.opendev.org/c/openstack/devstack/+/933704\n\nChange-Id: I792d6007725d0373daa9cf42218a6f28a316560c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/6350fdfc87dae68571e3344289a88d45a07f535a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/6350fdfc87dae68571e3344289a88d45a07f535a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
