)]}'
{"id":"openstack%2Fblazar-tempest-plugin~966362","triplet_id":"openstack%2Fblazar-tempest-plugin~master~If7c7be7fbf6d6ed78384088fc0a1f9873930ef85","project":"openstack/blazar-tempest-plugin","branch":"master","topic":"2024-1-last","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-11-13 14:56:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7c7be7fbf6d6ed78384088fc0a1f9873930ef85","subject":"Update stable branch jobs","status":"MERGED","created":"2025-11-06 22:15:30.000000000","updated":"2025-11-13 14:59:43.000000000","submitted":"2025-11-13 14:56:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966362-2024-1-last","meta_rev_id":"003d577090b93b0093ab91ad3868ee24e1f6e7b7","_number":966362,"virtual_id_number":966362,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-13 14:56:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-07 08:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-11-13 14:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-11-13 14:05:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 23:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-07 08:12:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-11-13 14:05:32.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"70d1568a9232605d238aa78d7e8ec1628419aa44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 22:15:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d179e4849b2b581172554fab21e8dcaa1565e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 23:04:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/316b43b8bd5543f0a0ca67f12fee30c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4484b506eecd4b59af13eba5e79565bf : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75acee4d68a44e6da8f6ca8982506d69 : SUCCESS in 4m 48s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c854a6d7e19c47d3ba386eb5c1b5126e : SUCCESS in 41m 27s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/300b73f336644881ba40077427d0fbfe : SUCCESS in 42m 25s\n- blazar-tempest-plugin-2025-2 https://zuul.opendev.org/t/openstack/build/30609a516c4c487387b33a8a8c0252bd : SUCCESS in 41m 59s\n- blazar-tempest-plugin-2025-1 https://zuul.opendev.org/t/openstack/build/9ce9e469680a4bb4a6d36c14c125d623 : FAILURE in 29m 10s\n- blazar-tempest-plugin-2024-2 https://zuul.opendev.org/t/openstack/build/f72c8ff4d2634169aa59379619abde31 : FAILURE in 20m 23s","accounts_in_message":[],"_revision_number":1},{"id":"aefc6311ae18a0f88eb19ad5e1e05aae74628563","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-07 03:43:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b3992c4d45958d530b28c4b7831785761a07ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 04:34:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10e397696c5b4367b6a53d11b3a3b521\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44611d4a4a6c47c7958d4449de53b443 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ee000a4c974aa5801cdfc6a259a430 : SUCCESS in 3m 46s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/3da5cc34cb0e429d83a3216c066fbbe4 : SUCCESS in 25m 11s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/54ef9b14d8f948bc84398b392963d4e2 : SUCCESS in 40m 42s\n- blazar-tempest-plugin-2025-2 https://zuul.opendev.org/t/openstack/build/72f47f2f2a1d4484974f0456b5a08e3d : SUCCESS in 25m 41s\n- blazar-tempest-plugin-2025-1 https://zuul.opendev.org/t/openstack/build/e9902ae030cd49198c51352deaa41069 : SUCCESS in 42m 43s\n- blazar-tempest-plugin-2024-2 https://zuul.opendev.org/t/openstack/build/5217801e11324bd5bf9ebcdd6daa3e65 : SUCCESS in 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"80dd7aaa23d2dc496a93770857c7967d292a2f9f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-07 08:12:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1268de72f60abf6b06c6b34d5fa01023a18528dd","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-13 14:05:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0fa36eb5596c464a5d9db9caf579d8ceb8cc804","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:06:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d606d181813c4b817ec23448df0e63bce78811b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:56:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f02f942605554671aafe0c05e38cb20b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b7af28f3cfa47139f862294725cad50 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7971f13806014de7b85e72ecd42e1da7 : SUCCESS in 5m 58s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/7edda408d82341209b4a48b089811fa0 : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6394e5d39f9ae20dc340bc923da8664dcfb4ef0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:56:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"003d577090b93b0093ab91ad3868ee24e1f6e7b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:59:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/753ae9021c154b1fa16fdb2d555dbb72\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c472c0697e946fa9dc2033ac2a86154 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e914a5abb1fd6f77e44e5374097255e3bef4fc6","revisions":{"8e914a5abb1fd6f77e44e5374097255e3bef4fc6":{"kind":"REWORK","_number":1,"created":"2025-11-06 22:15:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/62/966362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar-tempest-plugin","ref":"refs/changes/62/966362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/62/966362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/62/966362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/62/966362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar-tempest-plugin refs/changes/62/966362/1"}}},"commit":{"parents":[{"commit":"a834e47ef252750ca5e8e4b6bd1233fcaafa4322","subject":"Update stable branch jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/a834e47ef252750ca5e8e4b6bd1233fcaafa4322"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:39:37.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:39:37.000000000","tz":0},"subject":"Update stable branch jobs","message":"Update stable branch jobs\n\nAdd new stable/2025.2 job, remove unmaintained/2024.1\njobs.\n\nChange-Id: If7c7be7fbf6d6ed78384088fc0a1f9873930ef85\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/8e914a5abb1fd6f77e44e5374097255e3bef4fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar-tempest-plugin/commit/8e914a5abb1fd6f77e44e5374097255e3bef4fc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
