)]}'
{"id":"openstack%2Fblazar~585807","triplet_id":"openstack%2Fblazar~master~I44fd6faea2bb1d22a1578599f3858d7d24ff45b8","project":"openstack/blazar","branch":"master","topic":"soft-delete","attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-09-30 15:23:48.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-01-16 22:40:45.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I44fd6faea2bb1d22a1578599f3858d7d24ff45b8","subject":"Add soft delete to most Blazar models","status":"NEW","created":"2018-07-25 15:56:39.000000000","updated":"2024-01-16 22:40:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aa7dbc8cd586b6171302c15e6cd2f6905b8bd829","_number":585807,"virtual_id_number":585807,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-16 22:40:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-12 08:51:52.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2019-08-01 13:35:48.000000000","updated_by":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"reviewer":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"state":"REVIEWER"},{"updated":"2019-08-02 20:50:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"539214e255c8498b0310686da97877f34dfec97d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2018-07-25 15:56:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac846b71514128eb26da4bc871cdd90379a81daf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2018-07-25 16:10:02.000000000","message":"Patch Set 1:\n\nI would like to request feedback on this feature. I didn\u0027t implement soft delete for compute hosts and extra capabilities as that would require deep changes to the model to make each row unique even when deleted, as described in the commit message.\n\nOne way to do it would be to add a uuid field to each computehost that would be set by Blazar manager.\n\nAny thoughts?","accounts_in_message":[],"_revision_number":1},{"id":"85e1a5151710a594c8931f1db0e69e03404f78b7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2018-07-25 16:10:18.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"65350fb57ea651cbf34dda56c1b1d488e5a6db5c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2018-07-25 16:10:32.000000000","message":"Patch Set 1:\n\nPatch is ready but setting W-1 to gather feedback.","accounts_in_message":[],"_revision_number":1},{"id":"a6f43bc839b36b425867f7d4e77fddedee139f4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-25 18:41:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/07/585807/1/check/openstack-tox-pep8/3089089/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/07/585807/1/check/openstack-tox-py27/a4ad7a4/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/07/585807/1/check/openstack-tox-py35/5fc30fa/ : SUCCESS in 7m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/585807/1/check/build-openstack-sphinx-docs/c8492ad/html/ : SUCCESS in 5m 20s\n- openstack-tox-pylint http://logs.openstack.org/07/585807/1/check/openstack-tox-pylint/5a5eda1/ : SUCCESS in 6m 45s (non-voting)\n- blazar-devstack-dsvm http://logs.openstack.org/07/585807/1/check/blazar-devstack-dsvm/52d5bd6/ : FAILURE in 1h 02m 08s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/07/585807/1/check/blazar-devstack-dsvm-py35/bccc720/ : FAILURE in 1h 01m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/585807/1/check/openstack-tox-lower-constraints/3457fb1/ : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"247c2654eff67008813ab779036ce0e82f315432","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2018-07-27 13:01:06.000000000","message":"Patch Set 1:\n\nDevStack is blocking on running Alembic migrations, I am preparing a patch.","accounts_in_message":[],"_revision_number":1},{"id":"f1408ae3ce83144a45a84d7ceb37e13d6ca5bd75","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2018-07-27 14:28:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"21d9fd26a156d78cc2c402f8436c35e854d301b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-27 16:20:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/07/585807/2/check/openstack-tox-pep8/d2c384d/ : SUCCESS in 5m 58s\n- openstack-tox-py27 finger://ze06.openstack.org/f56f7d0e46764dfaa3811f630b2af040 : RETRY_LIMIT in 19s\n- openstack-tox-py35 finger://ze11.openstack.org/1c032aa8d844488b97946fc6903aa547 : RETRY_LIMIT in 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/585807/2/check/build-openstack-sphinx-docs/6b16438/html/ : SUCCESS in 6m 09s\n- openstack-tox-pylint http://logs.openstack.org/07/585807/2/check/openstack-tox-pylint/0edd14d/ : SUCCESS in 8m 39s (non-voting)\n- blazar-devstack-dsvm http://logs.openstack.org/07/585807/2/check/blazar-devstack-dsvm/0443559/ : SUCCESS in 40m 09s\n- blazar-devstack-dsvm-py35 finger://ze05.openstack.org/7f127c5f8ff64b09acd732474db1e15d : RETRY_LIMIT in 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/585807/2/check/openstack-tox-lower-constraints/1adfe40/ : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3b1c9b6c1b027637ff4dcf63f1d1d621a599a4d2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2018-07-30 11:30:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1807fa14a1cd65929e35fb71f6a865779685aac9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-30 12:14:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/07/585807/2/check/openstack-tox-pep8/8625ba2/ : SUCCESS in 9m 45s\n- openstack-tox-py27 http://logs.openstack.org/07/585807/2/check/openstack-tox-py27/80c99e6/ : SUCCESS in 9m 29s\n- openstack-tox-py35 http://logs.openstack.org/07/585807/2/check/openstack-tox-py35/e5c4853/ : SUCCESS in 5m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/585807/2/check/build-openstack-sphinx-docs/117230a/html/ : SUCCESS in 5m 10s\n- openstack-tox-pylint http://logs.openstack.org/07/585807/2/check/openstack-tox-pylint/195a03a/ : SUCCESS in 6m 25s (non-voting)\n- blazar-devstack-dsvm http://logs.openstack.org/07/585807/2/check/blazar-devstack-dsvm/45cadcc/ : SUCCESS in 37m 38s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/07/585807/2/check/blazar-devstack-dsvm-py35/d4f66c4/ : SUCCESS in 31m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/585807/2/check/openstack-tox-lower-constraints/00afb90/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"856b351ba60e1a666283b820ade3221e94956a74","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2018-07-31 09:51:09.000000000","message":"Patch Set 2: Workflow-1\n\nPatch is ready but some design points need to be discussed with the rest of team.","accounts_in_message":[],"_revision_number":2},{"id":"d85fcfc378f5934a9ff1d9497ca030191b7554d4","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2018-10-12 08:51:52.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nThanks for the first patch! Great work :-)\n\nLooks like this patch is for feedback, not for ready to go. I commented some points I noticed to care.","accounts_in_message":[],"_revision_number":2},{"id":"0a26858ba933d112677177a2c1a63e62b661d909","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-01 13:35:40.000000000","message":"Patch Set 2:\n\nNotes of required changes from the discussion at the Stein PTG:\n\n- Change the current patch series to \"resurrect\" soft-delete compute hosts when recreated by the operator\n- Implement Glance\u0027s approach for DB purge: https://docs.openstack.org/glance/latest/admin/db.html#purging-the-database","accounts_in_message":[],"_revision_number":2},{"id":"cd99fcfa97e1b6c9806de4a469f9e95a169216c9","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-08-02 19:29:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"465fe892fb254ee36b34ef96aa58f731f5bac3ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 20:04:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/07/585807/3/check/openstack-tox-lower-constraints/390d2c5/ : FAILURE in 4m 16s\n- openstack-tox-pep8 https://logs.opendev.org/07/585807/3/check/openstack-tox-pep8/2ad3087/ : SUCCESS in 3m 44s\n- openstack-tox-py27 https://logs.opendev.org/07/585807/3/check/openstack-tox-py27/0ab6f9a/ : FAILURE in 4m 17s\n- openstack-tox-py36 https://logs.opendev.org/07/585807/3/check/openstack-tox-py36/053a164/ : FAILURE in 3m 58s\n- openstack-tox-py37 https://logs.opendev.org/07/585807/3/check/openstack-tox-py37/68c9232/ : FAILURE in 4m 14s\n- openstack-tox-docs https://logs.opendev.org/07/585807/3/check/openstack-tox-docs/39fc710/html/ : SUCCESS in 3m 39s\n- blazar-devstack-dsvm https://logs.opendev.org/07/585807/3/check/blazar-devstack-dsvm/bf8ba3c/ : FAILURE in 28m 06s\n- blazar-devstack-dsvm-py35 https://logs.opendev.org/07/585807/3/check/blazar-devstack-dsvm-py35/5837072/ : FAILURE in 23m 13s\n- openstack-tox-pylint https://logs.opendev.org/07/585807/3/check/openstack-tox-pylint/182d9f0/ : SUCCESS in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"332be9da217c6a51f284fd406a1fc00b84e5cb4b","author":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"date":"2019-08-02 20:21:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"fa663952c5ac125d743fe8d7e96b9fe351586d78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 20:50:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/07/585807/4/check/openstack-tox-lower-constraints/f23ec54/ : FAILURE in 4m 11s\n- openstack-tox-pep8 https://logs.opendev.org/07/585807/4/check/openstack-tox-pep8/47031da/ : SUCCESS in 3m 41s\n- openstack-tox-py27 https://logs.opendev.org/07/585807/4/check/openstack-tox-py27/15ae2c5/ : FAILURE in 4m 11s\n- openstack-tox-py36 https://logs.opendev.org/07/585807/4/check/openstack-tox-py36/b3dc551/ : FAILURE in 4m 24s\n- openstack-tox-py37 https://logs.opendev.org/07/585807/4/check/openstack-tox-py37/f66921c/ : FAILURE in 4m 34s\n- openstack-tox-docs https://logs.opendev.org/07/585807/4/check/openstack-tox-docs/1817191/html/ : SUCCESS in 4m 19s\n- blazar-devstack-dsvm https://logs.opendev.org/07/585807/4/check/blazar-devstack-dsvm/ca8142f/ : FAILURE in 24m 31s\n- blazar-devstack-dsvm-py35 https://logs.opendev.org/07/585807/4/check/blazar-devstack-dsvm-py35/3bbb712/ : FAILURE in 23m 20s\n- openstack-tox-pylint https://logs.opendev.org/07/585807/4/check/openstack-tox-pylint/fc977a8/ : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"08704e1d7bc8981344c9d558dfcccae16ecc8608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-09-30 14:41:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f4042c64f2c6d9108243a76cda9bf6def61ff68d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 15:23:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea20a03dd3134edebe1f24fb6b22ef3a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0398e2ad81ac47b49669d825e04aebee : FAILURE in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4dc56dbd1a49af91f0929e24f1fd6e : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b0ed57d65f447bf9ff1dc6d88dc9d48 : FAILURE in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a129ec16d724b9a9e807054bd9db38e : FAILURE in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c150f04aa4b44a44a5f3507224a9d623 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b679003f62c46a5b280c1629c2adb5d : SUCCESS in 5m 14s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/7aac7deb373b4aa0b26883131dd1ff71 : FAILURE in 24m 58s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/dfb202faab81437c8d2ba8527dfafeff : FAILURE in 31m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14d99a2747e24ec2a1601bdf2a2ee081 : FAILURE in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c2337b04160ff3bcb8f8fe4bb78bf25f1956e9a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-10-14 14:31:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b164ec0597c8113b5d097fc992e9aeb011f7fc4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 15:03:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca653a57110b4fb5a514c1a8d2559330\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d63c83d9cf724bdf81de292ea8d9a003 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73fbb6be587458db0e651536355463e : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/228994beaa4c474e97b8357022921e8d : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0695c11dd3154419997e7d041d195101 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb9b2d143fb4fc2bb84c585d73cd08a : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4060242a6c35464ca0448d27f72cdacf : SUCCESS in 3m 27s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/2d50d2fdb8004884a9782f0cbf8a38b8 : FAILURE in 27m 02s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/53991c79c4da49758a279a6ed68a7d9b : FAILURE in 25m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5dd42db850e2426ca5a6364c09c1aacb : FAILURE in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4f43b72543f5e39e5c872ddf73d81dd439abb0e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2022-10-17 08:27:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3dcf3ba2c160757c6ea21ceb4198fd2f76ceeb9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 09:06:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/626617ddf2a24f70b99a28df183a3972\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61463fc3dd3443df82bc2084bbe43dfe : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5ba19769e542c1b49ab206c8900326 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80f4f35bbdf742d78f86185552d581ad : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7888320dc88741199f5f03620e2c7e3d : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdbacfe661dd45439f2be8a371880cb4 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/115258a1efca4b6ca1c1482ff67a7da1 : SUCCESS in 3m 52s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/69debcf424694a7d875c41213012957f : SUCCESS in 31m 52s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a72bda199f62470f8d4af22c31e95446 : SUCCESS in 37m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c6dcaa162d834a46940ecffe8d79630e : FAILURE in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e089deaa56813c8455486aca80bfa85e8dd2507c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-01-16 21:44:26.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"aa7dbc8cd586b6171302c15e6cd2f6905b8bd829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 22:40:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03b504cf5e054ad7abc814690301c51d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c790d4bc1d224b478333efd2e15749e2 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e47a23f3bf644b3da58204f6fa330778 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e89301b663f4a37a36fd98f8c45dd71 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b044bc2f65c9444abc7426f8bc8a53d2 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678e2b951aba4280bb013ad84abe749d : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4181fe205bb44a9da0484af49118c92e : SUCCESS in 3m 25s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ada3cafbaed54b4daacba2548cfa22c4 : SUCCESS in 46m 25s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/05987e8e1d38422d867bce7cd2aeb187 : SUCCESS in 49m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b393174f3c1544bc9965df658444fa7e : FAILURE in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"730812e37d259cc9d84bdf075710ed5cecb8182b","revisions":{"4716a62a56a488a04d0eeaaf03cb35854738100b":{"kind":"REWORK","_number":1,"created":"2018-07-25 15:56:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/585807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/585807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/585807/1"}}},"commit":{"parents":[{"commit":"9f97e34e6a6b8a290255b7816bb1822b61b8deb5","subject":"Fix default values provided as keyword arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/9f97e34e6a6b8a290255b7816bb1822b61b8deb5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-25 13:59:47.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-25 15:28:56.000000000","tz":60},"subject":"Add soft delete to most Blazar models","message":"Add soft delete to most Blazar models\n\nThis patch creates a SoftDeleteMixinWithUuid class from which most\nBlazar models inherit and adapts the rest of the code to soft delete\nrows. An Alembic migration script is included to add `deleted` and\n`deleted_at` columns to all tables.\n\nCompute hosts and extra capabilities are not given soft delete\ncapabilities: compute hosts primary keys match Nova compute node IDs,\nwhich could create conflicts if a host is deleted and re-created.\nAccordingly, the Alembic migration script removes the foreign key\nrelationship between computehost_allocations and computehosts, since\ncompute hosts can be deleted from the database while soft deleted\nallocations are kept.\n\nChange-Id: I44fd6faea2bb1d22a1578599f3858d7d24ff45b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4716a62a56a488a04d0eeaaf03cb35854738100b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4716a62a56a488a04d0eeaaf03cb35854738100b"}]},"branch":"refs/heads/master"},"c91381f48a24af5648e1f41490e97ea326d87fe9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-07-27 14:28:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/585807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/585807/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/585807/2"}}},"commit":{"parents":[{"commit":"b6eb8ee6bec6142254d077090ac5b3e43fd4fc75","subject":"Fix default values provided as keyword arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b6eb8ee6bec6142254d077090ac5b3e43fd4fc75"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-25 13:59:47.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-27 14:27:38.000000000","tz":60},"subject":"Add soft delete to most Blazar models","message":"Add soft delete to most Blazar models\n\nThis patch creates a SoftDeleteMixinWithUuid class from which most\nBlazar models inherit and adapts the rest of the code to soft delete\nrows. An Alembic migration script is included to add `deleted` and\n`deleted_at` columns to all tables.\n\nCompute hosts and extra capabilities are not given soft delete\ncapabilities: compute hosts primary keys match Nova compute node IDs,\nwhich could create conflicts if a host is deleted and re-created.\nAccordingly, the Alembic migration script removes the foreign key\nrelationship between computehost_allocations and computehosts, since\ncompute hosts can be deleted from the database while soft deleted\nallocations are kept.\n\nChange-Id: I44fd6faea2bb1d22a1578599f3858d7d24ff45b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c91381f48a24af5648e1f41490e97ea326d87fe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c91381f48a24af5648e1f41490e97ea326d87fe9"}]},"branch":"refs/heads/master"},"ab8271ecfe970e8718c8db7b5c249253a82241e6":{"kind":"REWORK","_number":3,"created":"2019-08-02 19:29:57.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/07/585807/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/585807/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/585807/3"}}},"commit":{"parents":[{"commit":"5123399a917413c2cc9f621987d0721fc15ca737","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5123399a917413c2cc9f621987d0721fc15ca737"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-25 13:59:47.000000000","tz":60},"committer":{"name":"jakecoll","email":"jakecoll@uchicago.edu","date":"2019-08-02 19:29:35.000000000","tz":-300},"subject":"Add soft delete to most Blazar models","message":"Add soft delete to most Blazar models\n\nThis patch creates a SoftDeleteMixinWithUuid class from which most\nBlazar models inherit and adapts the rest of the code to soft delete\nrows. An Alembic migration script is included to add `deleted` and\n`deleted_at` columns to all tables.\n\nCompute hosts and extra capabilities are not given soft delete\ncapabilities: compute hosts primary keys match Nova compute node IDs,\nwhich could create conflicts if a host is deleted and re-created.\nAccordingly, the Alembic migration script removes the foreign key\nrelationship between computehost_allocations and computehosts, since\ncompute hosts can be deleted from the database while soft deleted\nallocations are kept.\n\nChange-Id: I44fd6faea2bb1d22a1578599f3858d7d24ff45b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ab8271ecfe970e8718c8db7b5c249253a82241e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ab8271ecfe970e8718c8db7b5c249253a82241e6"}]},"branch":"refs/heads/master"},"11e90a13818d3c0a0183cbd4a8eccac1e88a4454":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-02 20:21:32.000000000","uploader":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"ref":"refs/changes/07/585807/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/585807/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/585807/4"}}},"commit":{"parents":[{"commit":"64a1d26aaf0368af4ea4773b41520a0802786fb3","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/64a1d26aaf0368af4ea4773b41520a0802786fb3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-25 13:59:47.000000000","tz":60},"committer":{"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","date":"2019-08-02 20:21:32.000000000","tz":0},"subject":"Add soft delete to most Blazar models","message":"Add soft delete to most Blazar models\n\nThis patch creates a SoftDeleteMixinWithUuid class from which most\nBlazar models inherit and adapts the rest of the code to soft delete\nrows. An Alembic migration script is included to add `deleted` and\n`deleted_at` columns to all tables.\n\nCompute hosts and extra capabilities are not given soft delete\ncapabilities: compute hosts primary keys match Nova compute node IDs,\nwhich could create conflicts if a host is deleted and re-created.\nAccordingly, the Alembic migration script removes the foreign key\nrelationship between computehost_allocations and computehosts, since\ncompute hosts can be deleted from the database while soft deleted\nallocations are kept.\n\nChange-Id: I44fd6faea2bb1d22a1578599f3858d7d24ff45b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/11e90a13818d3c0a0183cbd4a8eccac1e88a4454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/11e90a13818d3c0a0183cbd4a8eccac1e88a4454"}]},"branch":"refs/heads/master"},"a61d9b8582aab1b65767e20ae97b9800786bbf7a":{"kind":"REWORK","_number":5,"created":"2022-09-30 14:41:12.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/07/585807/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/585807/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/585807/5"}}},"commit":{"parents":[{"commit":"da81c207098738e62b85d85bf09f0f3dfa8d74b8","subject":"Update copyright date for release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da81c207098738e62b85d85bf09f0f3dfa8d74b8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-25 13:59:47.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-09-30 14:37:34.000000000","tz":0},"subject":"Add soft delete to most Blazar models","message":"Add soft delete to most Blazar models\n\nThis patch creates a SoftDeleteMixinWithUuid class from which most\nBlazar models inherit and adapts the rest of the code to soft delete\nrows. An Alembic migration script is included to add `deleted` and\n`deleted_at` columns to all tables.\n\nCompute hosts and extra capabilities are not given soft delete\ncapabilities: compute hosts primary keys match Nova compute node IDs,\nwhich could create conflicts if a host is deleted and re-created.\nAccordingly, the Alembic migration script removes the foreign key\nrelationship between computehost_allocations and computehosts, since\ncompute hosts can be deleted from the database while soft deleted\nallocations are kept.\n\nThe database can be pruned of soft-deleted entries with the command\n``blazar-db-manage purge --age_in_days`` which will remove any rows\nthat have been soft deleted that are older than the given age.\n`age_in_days` defaults to 30.\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I44fd6faea2bb1d22a1578599f3858d7d24ff45b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a61d9b8582aab1b65767e20ae97b9800786bbf7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a61d9b8582aab1b65767e20ae97b9800786bbf7a"}]},"branch":"refs/heads/master"},"c6a74719fdcec5d79d6cbee8fd8fef5a7fae45a7":{"kind":"REWORK","_number":6,"created":"2022-10-14 14:31:37.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/07/585807/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/585807/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/585807/6"}}},"commit":{"parents":[{"commit":"da81c207098738e62b85d85bf09f0f3dfa8d74b8","subject":"Update copyright date for release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da81c207098738e62b85d85bf09f0f3dfa8d74b8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-25 13:59:47.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-10-14 14:30:50.000000000","tz":60},"subject":"Add soft delete to most Blazar models","message":"Add soft delete to most Blazar models\n\nThis patch creates a SoftDeleteMixinWithUuid class from which most\nBlazar models inherit and adapts the rest of the code to soft delete\nrows. An Alembic migration script is included to add `deleted` and\n`deleted_at` columns to all tables.\n\nCompute hosts and extra capabilities are not given soft delete\ncapabilities: compute hosts primary keys match Nova compute node IDs,\nwhich could create conflicts if a host is deleted and re-created.\nAccordingly, the Alembic migration script removes the foreign key\nrelationship between computehost_allocations and computehosts, since\ncompute hosts can be deleted from the database while soft deleted\nallocations are kept.\n\nThe database can be pruned of soft-deleted entries with the command\n``blazar-db-manage purge --age_in_days`` which will remove any rows\nthat have been soft deleted that are older than the given age.\n`age_in_days` defaults to 30.\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I44fd6faea2bb1d22a1578599f3858d7d24ff45b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c6a74719fdcec5d79d6cbee8fd8fef5a7fae45a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c6a74719fdcec5d79d6cbee8fd8fef5a7fae45a7"}]},"branch":"refs/heads/master"},"42737075d69b733101f0e641cf8ad3e50a0b984e":{"kind":"REWORK","_number":7,"created":"2022-10-17 08:27:20.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/07/585807/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/585807/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/585807/7"}}},"commit":{"parents":[{"commit":"da81c207098738e62b85d85bf09f0f3dfa8d74b8","subject":"Update copyright date for release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da81c207098738e62b85d85bf09f0f3dfa8d74b8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-25 13:59:47.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2022-10-17 08:27:07.000000000","tz":60},"subject":"Add soft delete to most Blazar models","message":"Add soft delete to most Blazar models\n\nThis patch creates a SoftDeleteMixinWithUuid class from which most\nBlazar models inherit and adapts the rest of the code to soft delete\nrows. An Alembic migration script is included to add `deleted` and\n`deleted_at` columns to all tables.\n\nCompute hosts and extra capabilities are not given soft delete\ncapabilities: compute hosts primary keys match Nova compute node IDs,\nwhich could create conflicts if a host is deleted and re-created.\nAccordingly, the Alembic migration script removes the foreign key\nrelationship between computehost_allocations and computehosts, since\ncompute hosts can be deleted from the database while soft deleted\nallocations are kept.\n\nThe database can be pruned of soft-deleted entries with the command\n``blazar-db-manage purge --age_in_days`` which will remove any rows\nthat have been soft deleted that are older than the given age.\n`age_in_days` defaults to 30.\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I44fd6faea2bb1d22a1578599f3858d7d24ff45b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/42737075d69b733101f0e641cf8ad3e50a0b984e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/42737075d69b733101f0e641cf8ad3e50a0b984e"}]},"branch":"refs/heads/master"},"730812e37d259cc9d84bdf075710ed5cecb8182b":{"kind":"REWORK","_number":8,"created":"2024-01-16 21:44:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/585807/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/07/585807/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/07/585807/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/07/585807/8"}}},"commit":{"parents":[{"commit":"b0dbd0a9299cab3451c1b3abdd35f6b5e76ca14a","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b0dbd0a9299cab3451c1b3abdd35f6b5e76ca14a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2018-07-25 13:59:47.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-01-16 21:43:58.000000000","tz":60},"subject":"Add soft delete to most Blazar models","message":"Add soft delete to most Blazar models\n\nThis patch creates a SoftDeleteMixinWithUuid class from which most\nBlazar models inherit and adapts the rest of the code to soft delete\nrows. An Alembic migration script is included to add `deleted` and\n`deleted_at` columns to all tables.\n\nCompute hosts and extra capabilities are not given soft delete\ncapabilities: compute hosts primary keys match Nova compute node IDs,\nwhich could create conflicts if a host is deleted and re-created.\nAccordingly, the Alembic migration script removes the foreign key\nrelationship between computehost_allocations and computehosts, since\ncompute hosts can be deleted from the database while soft deleted\nallocations are kept.\n\nThe database can be pruned of soft-deleted entries with the command\n`blazar-db-manage purge --age_in_days` which will remove any rows that\nhave been soft deleted that are older than the given age. `age_in_days`\ndefaults to 30.\n\nCo-Authored-By: Matt Crees \u003cmattc@stackhpc.com\u003e\n\nChange-Id: I44fd6faea2bb1d22a1578599f3858d7d24ff45b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/730812e37d259cc9d84bdf075710ed5cecb8182b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/730812e37d259cc9d84bdf075710ed5cecb8182b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
