)]}'
{"id":"openstack%2Fblazar~630026","triplet_id":"openstack%2Fblazar~master~I70e35ac7a524afb04077da5978c7563dec120639","project":"openstack/blazar","branch":"master","topic":"bug/1821576","hashtags":[],"change_id":"I70e35ac7a524afb04077da5978c7563dec120639","subject":"Lease creation/update should fail if amount is 0","status":"MERGED","created":"2019-01-11 05:18:07.000000000","updated":"2019-09-16 17:14:38.000000000","submitted":"2019-09-16 17:12:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"630026-1568653977217-354c625b","meta_rev_id":"08bff464070e3bf2aabdf9135d2f4da9b14eeae7","_number":630026,"virtual_id_number":630026,"owner":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-09-16 17:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"value":0,"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-16 16:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"value":0,"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-16 16:14:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"value":0,"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"value":0,"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-19 06:42:39.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2019-06-03 11:37:08.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-09-16 16:14:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-09-16 17:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd13a266dcc1908557522d957d6788f1b06c5c92","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-01-11 05:18:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf5d1275b981a15e7de12d2ad5a991e7a1f5b5dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 06:13:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/26/630026/1/check/openstack-tox-lower-constraints/0c33bbd/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 http://logs.openstack.org/26/630026/1/check/openstack-tox-pep8/fd9d37d/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/26/630026/1/check/openstack-tox-py27/1659725/ : SUCCESS in 4m 21s\n- openstack-tox-py35 http://logs.openstack.org/26/630026/1/check/openstack-tox-py35/ce87df5/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/26/630026/1/check/openstack-tox-py36/4a85aac/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/26/630026/1/check/openstack-tox-docs/43e7d9e/html/ : SUCCESS in 4m 10s\n- blazar-devstack-dsvm http://logs.openstack.org/26/630026/1/check/blazar-devstack-dsvm/c27f767/ : SUCCESS in 48m 20s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/26/630026/1/check/blazar-devstack-dsvm-py35/de1eed0/ : SUCCESS in 44m 55s\n- openstack-tox-pylint http://logs.openstack.org/26/630026/1/check/openstack-tox-pylint/790730d/ : SUCCESS in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cec93a86c5d9c284912e54bc41487b9c3524af75","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2019-01-29 08:19:23.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for the patch.\n\nI\u0027m okay to make the constraints that minimum value of amount or max and min value is 1.  Could you also add same validation to update_reservation() in both the instance plugin and the host plugin, and update Blazar\u0027s API references?","accounts_in_message":[],"_revision_number":1},{"id":"812db81af96633b9a8967ff0f2793c4fa2478d3a","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-02-13 07:29:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa616899c42a86989e77a085cf8f6bed2bdb57ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 08:15:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/26/630026/2/check/build-openstack-api-ref/61a0b36/html/ : SUCCESS in 4m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/630026/2/check/openstack-tox-lower-constraints/fe82281/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/26/630026/2/check/openstack-tox-pep8/986a420/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/26/630026/2/check/openstack-tox-py27/0dfa288/ : SUCCESS in 6m 51s\n- openstack-tox-py35 http://logs.openstack.org/26/630026/2/check/openstack-tox-py35/4e63b43/ : SUCCESS in 6m 56s\n- openstack-tox-py36 http://logs.openstack.org/26/630026/2/check/openstack-tox-py36/24315f9/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/26/630026/2/check/openstack-tox-docs/d87288c/html/ : SUCCESS in 4m 05s\n- blazar-devstack-dsvm http://logs.openstack.org/26/630026/2/check/blazar-devstack-dsvm/2c4ea22/ : SUCCESS in 37m 37s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/26/630026/2/check/blazar-devstack-dsvm-py35/e350a8a/ : SUCCESS in 40m 13s\n- openstack-tox-pylint http://logs.openstack.org/26/630026/2/check/openstack-tox-pylint/67cf67b/ : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09e440a55ce1fe2966e5f027a74d23e6173d8fd1","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2019-02-26 05:18:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nThanks for the update. The change looks good to me. Just one style comment.","accounts_in_message":[],"_revision_number":2},{"id":"20383725761c66726b26f2d61d8e9b4416790411","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-02-26 08:47:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eb3aeb637ca92fe594e5f2fc1493e5820e6a85b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 09:32:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/26/630026/3/check/build-openstack-api-ref/569c74a/html/ : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/630026/3/check/openstack-tox-lower-constraints/15122a5/ : SUCCESS in 6m 18s\n- openstack-tox-pep8 http://logs.openstack.org/26/630026/3/check/openstack-tox-pep8/e1a63b9/ : SUCCESS in 5m 56s\n- openstack-tox-py27 http://logs.openstack.org/26/630026/3/check/openstack-tox-py27/29f6e2c/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/26/630026/3/check/openstack-tox-py35/7b54daa/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/26/630026/3/check/openstack-tox-py36/cc57cbc/ : SUCCESS in 5m 10s\n- openstack-tox-py37 http://logs.openstack.org/26/630026/3/check/openstack-tox-py37/9e5b52b/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/26/630026/3/check/openstack-tox-docs/79ff8b1/html/ : SUCCESS in 6m 20s\n- blazar-devstack-dsvm http://logs.openstack.org/26/630026/3/check/blazar-devstack-dsvm/69cb0e6/ : SUCCESS in 36m 47s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/26/630026/3/check/blazar-devstack-dsvm-py35/f657296/ : FAILURE in 40m 49s\n- openstack-tox-pylint http://logs.openstack.org/26/630026/3/check/openstack-tox-pylint/5c611ac/ : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4e6bd157adb04198feb12f3359ddab6c89ff52f7","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-02-26 09:34:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a06a02e332287fd477abe4a071c9900fbd595fcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 10:16:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/26/630026/3/check/build-openstack-api-ref/5c46424/html/ : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/630026/3/check/openstack-tox-lower-constraints/437688a/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/26/630026/3/check/openstack-tox-pep8/2896892/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/26/630026/3/check/openstack-tox-py27/95cb149/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/26/630026/3/check/openstack-tox-py35/924c0d2/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/26/630026/3/check/openstack-tox-py36/2854a15/ : SUCCESS in 5m 27s\n- openstack-tox-py37 http://logs.openstack.org/26/630026/3/check/openstack-tox-py37/34c2dbf/ : SUCCESS in 6m 12s\n- openstack-tox-docs http://logs.openstack.org/26/630026/3/check/openstack-tox-docs/fc1a2fd/html/ : SUCCESS in 5m 24s\n- blazar-devstack-dsvm http://logs.openstack.org/26/630026/3/check/blazar-devstack-dsvm/7a85f19/ : SUCCESS in 37m 09s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/26/630026/3/check/blazar-devstack-dsvm-py35/31d2198/ : SUCCESS in 36m 24s\n- openstack-tox-pylint http://logs.openstack.org/26/630026/3/check/openstack-tox-pylint/39edbf2/ : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e77dda1c882669c453fe7709aaafd5bc806e3fde","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2019-03-19 06:42:39.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for the update. Need resolving the conflict.","accounts_in_message":[],"_revision_number":3},{"id":"6138b5d0fa8419cda09cddeccc06cd9bcd42f115","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-03-19 09:35:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fbfa70b91c8f270baf8dd48fd296c141f5370120","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-19 10:29:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/26/630026/4/check/build-openstack-api-ref/06c49dd/html/ : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/630026/4/check/openstack-tox-lower-constraints/6af4f7f/ : SUCCESS in 5m 34s\n- openstack-tox-pep8 http://logs.openstack.org/26/630026/4/check/openstack-tox-pep8/88effd6/ : SUCCESS in 5m 51s\n- openstack-tox-py27 http://logs.openstack.org/26/630026/4/check/openstack-tox-py27/9aead9c/ : SUCCESS in 5m 55s\n- openstack-tox-py35 http://logs.openstack.org/26/630026/4/check/openstack-tox-py35/ef877c1/ : SUCCESS in 5m 59s\n- openstack-tox-py36 http://logs.openstack.org/26/630026/4/check/openstack-tox-py36/01a3398/ : SUCCESS in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/26/630026/4/check/openstack-tox-py37/9a3435b/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/26/630026/4/check/openstack-tox-docs/f8dbacb/html/ : SUCCESS in 5m 51s\n- blazar-devstack-dsvm http://logs.openstack.org/26/630026/4/check/blazar-devstack-dsvm/0c9bb56/ : SUCCESS in 44m 42s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/26/630026/4/check/blazar-devstack-dsvm-py35/1366031/ : SUCCESS in 41m 53s\n- openstack-tox-pylint http://logs.openstack.org/26/630026/4/check/openstack-tox-pylint/04257ad/ : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"329160e60acb69c02875ae831080be2c43592d17","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-03-20 23:05:05.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9a45d5b7d6509adf20da847c7096f86f9a9f17a4","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-28 04:22:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c5b9703f281c743fca2c015877767ca6004d398b","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-03-28 11:34:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"02d3477ca62094bdda396be25892f9bc0d48d626","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 12:28:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/26/630026/5/check/build-openstack-api-ref/0056234/html/ : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/630026/5/check/openstack-tox-lower-constraints/8b88e91/ : SUCCESS in 6m 39s\n- openstack-tox-pep8 http://logs.openstack.org/26/630026/5/check/openstack-tox-pep8/e226731/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/26/630026/5/check/openstack-tox-py27/b121406/ : SUCCESS in 7m 45s\n- openstack-tox-py35 http://logs.openstack.org/26/630026/5/check/openstack-tox-py35/468431e/ : SUCCESS in 8m 20s\n- openstack-tox-py36 http://logs.openstack.org/26/630026/5/check/openstack-tox-py36/2108e6a/ : SUCCESS in 7m 47s\n- openstack-tox-py37 http://logs.openstack.org/26/630026/5/check/openstack-tox-py37/20c05e5/ : SUCCESS in 8m 27s\n- openstack-tox-docs http://logs.openstack.org/26/630026/5/check/openstack-tox-docs/1b216ea/html/ : SUCCESS in 6m 58s\n- blazar-devstack-dsvm http://logs.openstack.org/26/630026/5/check/blazar-devstack-dsvm/d349161/ : SUCCESS in 39m 58s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/26/630026/5/check/blazar-devstack-dsvm-py35/85210f8/ : SUCCESS in 36m 15s\n- openstack-tox-pylint http://logs.openstack.org/26/630026/5/check/openstack-tox-pylint/376f249/ : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"caf3451a9d84f87fa99ee9733a35975ee09d2e31","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-04-03 08:29:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fc0d9232bbd466c7fe949dd7cbe5f5293b6cade5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 09:13:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/26/630026/6/check/build-openstack-api-ref/4954205/html/ : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/630026/6/check/openstack-tox-lower-constraints/fad6367/ : SUCCESS in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/26/630026/6/check/openstack-tox-pep8/faeb721/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/26/630026/6/check/openstack-tox-py27/95de635/ : SUCCESS in 6m 02s\n- openstack-tox-py35 http://logs.openstack.org/26/630026/6/check/openstack-tox-py35/c755da3/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/26/630026/6/check/openstack-tox-py36/260315e/ : SUCCESS in 5m 32s\n- openstack-tox-py37 http://logs.openstack.org/26/630026/6/check/openstack-tox-py37/7c0afde/ : SUCCESS in 8m 16s\n- openstack-tox-docs http://logs.openstack.org/26/630026/6/check/openstack-tox-docs/2ec7c54/html/ : SUCCESS in 5m 14s\n- blazar-devstack-dsvm http://logs.openstack.org/26/630026/6/check/blazar-devstack-dsvm/ed01685/ : SUCCESS in 37m 23s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/26/630026/6/check/blazar-devstack-dsvm-py35/151d04b/ : SUCCESS in 38m 47s\n- openstack-tox-pylint http://logs.openstack.org/26/630026/6/check/openstack-tox-pylint/13c3586/ : SUCCESS in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b95d07c5ce3e2444d061990f7ae914ad0a20d08f","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-04-10 06:53:46.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"91468a5ce885e4ae749ce7c8b6a7623fc0f5d774","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-30 15:46:01.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"56e91567f6fb53f8aceb73e0d4f027727b81ddb3","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-06-03 06:48:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0126f8ce93a46dfb41baa81c447a7865c5ff3110","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-06-03 11:37:08.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"30f98ae9d2c3a37ddbc77fe0fd56573c1bba8012","author":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"date":"2019-06-06 09:05:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"441d505e7f231982c04f6b445fc86bc1d5c6daff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:01:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/26/630026/7/check/build-openstack-api-ref/d91ab7e/html/ : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/26/630026/7/check/openstack-tox-lower-constraints/d2f2463/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/26/630026/7/check/openstack-tox-pep8/79fad40/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/26/630026/7/check/openstack-tox-py27/d948ae3/ : SUCCESS in 6m 10s\n- openstack-tox-py35 http://logs.openstack.org/26/630026/7/check/openstack-tox-py35/f4f8e11/ : SUCCESS in 6m 22s\n- openstack-tox-py36 http://logs.openstack.org/26/630026/7/check/openstack-tox-py36/a3190c6/ : SUCCESS in 5m 49s\n- openstack-tox-py37 http://logs.openstack.org/26/630026/7/check/openstack-tox-py37/6c88225/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/26/630026/7/check/openstack-tox-docs/ce6c2d6/html/ : SUCCESS in 4m 25s\n- blazar-devstack-dsvm http://logs.openstack.org/26/630026/7/check/blazar-devstack-dsvm/8a6ff40/ : SUCCESS in 37m 22s\n- blazar-devstack-dsvm-py35 http://logs.openstack.org/26/630026/7/check/blazar-devstack-dsvm-py35/dcf6243/ : SUCCESS in 38m 32s\n- openstack-tox-pylint http://logs.openstack.org/26/630026/7/check/openstack-tox-pylint/f61e25f/ : SUCCESS in 6m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2e5ff0dc9023691e03dc59b1329e8c10c1f32d2e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-16 16:14:19.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"affacf1f8aac7820d90791a69b9ab0c8ae28cf57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 16:14:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1fa79f3a4f435e5406000c1184ab9059b7809778","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-16 16:14:38.000000000","message":"Patch Set 7:\n\nReusing earlier +2 from Tetsuro, minor changes since.","accounts_in_message":[],"_revision_number":7},{"id":"0ba68ff9d560454897465b4debfb407a3854915e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 17:12:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/00b91f4ed1bb42f8a76969deb7d3a613 : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad47edd121d04b4da16295e960fada40 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d009a5ec90f7403da7002968a0e2a2c6 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcea97f849d546beba2bdfc63f5fcad9 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3266a0b1a7864c3daba29a14f79b2c2d : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e79d2e4f07e412f80387b67d05cca67 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83bcea58add48008c291926d5dcf831 : SUCCESS in 6m 07s\n- blazar-devstack-dsvm https://zuul.opendev.org/t/openstack/build/f8077e92d7044cd8af56d773cfa74897 : SUCCESS in 49m 23s\n- blazar-devstack-dsvm-py35 https://zuul.opendev.org/t/openstack/build/c1d74a32318d48968e3160ccee607dec : SUCCESS in 42m 55s","accounts_in_message":[],"_revision_number":7},{"id":"4f6393ea0fc9f5c296d99910cf474d1d59d85e10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 17:12:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d7543ea34422af18dbf1001bc16f6f57a03bee69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 17:14:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6cbb70a881d940dcba76edbafdc56162 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003fb71f8de1497281dc5305acc28a0d : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"fa6bb0b5b2976906214b7f7f4f4aa5cb24234736","revisions":{"f2ff867f58cef71bd8160da9baa182431986cf97":{"kind":"REWORK","_number":1,"created":"2019-01-11 05:18:07.000000000","uploader":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"ref":"refs/changes/26/630026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/26/630026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/26/630026/1"}}},"commit":{"parents":[{"commit":"b7b600efc2ce1960e5b7b4626fcc10cf29def699","subject":"Merge \"Change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b7b600efc2ce1960e5b7b4626fcc10cf29def699"}]}],"author":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2018-12-21 14:00:18.000000000","tz":0},"committer":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2019-01-11 04:48:54.000000000","tz":0},"subject":"Lease creation should fail if amount is 0","message":"Lease creation should fail if amount is 0\n\nLease is getting created for resource_type virtual:instance\nwith amount\u003d0.\nThis patch doesn\u0027t allow creation of lease with amount\u003d0,\nit will raise 400 BadRequest.\n\nLease creation for resource_type physical:host with\nmin\u003d0 or max\u003d0 raises 500 Internal Server Error.\nThis patch doesn\u0027t allow creation of lease with\nmin\u003d0 or max\u003d0, it will raise 400 BadRequest.\n\nAPIImpact:\n1)Return 400 instead of 201 when lease is created with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n2)Return 400 instead of 500 when lease is created with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n\nCloses-Bug: #1807101\n\nChange-Id: I70e35ac7a524afb04077da5978c7563dec120639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f2ff867f58cef71bd8160da9baa182431986cf97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f2ff867f58cef71bd8160da9baa182431986cf97"}]},"branch":"refs/heads/master"},"f2afbfae5d59e8552b8e8d8807a0fa45aa004ad9":{"kind":"REWORK","_number":2,"created":"2019-02-13 07:29:14.000000000","uploader":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"ref":"refs/changes/26/630026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/26/630026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/26/630026/2"}}},"commit":{"parents":[{"commit":"b7b600efc2ce1960e5b7b4626fcc10cf29def699","subject":"Merge \"Change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b7b600efc2ce1960e5b7b4626fcc10cf29def699"}]}],"author":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2018-12-21 14:00:18.000000000","tz":0},"committer":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2019-02-12 13:37:29.000000000","tz":0},"subject":"Lease creation/updation should fail if amount is 0","message":"Lease creation/updation should fail if amount is 0\n\nLease is getting created and updated for resource_type\nvirtual:instance with amount\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with amount\u003d0, it will raise 400 BadRequest.\n\nLease creation for resource_type physical:host with\nmin\u003d0 or max\u003d0 raises 500 Internal Server Error\nand lease is updated for resource_type physical:host\nwith min\u003d0 or max\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with min\u003d0 or max\u003d0, it will raise 400 BadRequest.\n\nAPIImpact:\n1)Return 400 instead of 201 when lease is created with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n2)Return 400 instead of 500 when lease is created with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n3)Return 400 instead of 200 when lease is updated with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n4)Return 400 instead of 200 when lease is updated with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n\nCloses-Bug: #1807101\n\nChange-Id: I70e35ac7a524afb04077da5978c7563dec120639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f2afbfae5d59e8552b8e8d8807a0fa45aa004ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/f2afbfae5d59e8552b8e8d8807a0fa45aa004ad9"}]},"branch":"refs/heads/master"},"e2f0f58665636c46ac06730bf0fd6789fce8d2c4":{"kind":"REWORK","_number":3,"created":"2019-02-26 08:47:33.000000000","uploader":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"ref":"refs/changes/26/630026/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/26/630026/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/26/630026/3"}}},"commit":{"parents":[{"commit":"b7b600efc2ce1960e5b7b4626fcc10cf29def699","subject":"Merge \"Change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b7b600efc2ce1960e5b7b4626fcc10cf29def699"}]}],"author":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2018-12-21 14:00:18.000000000","tz":0},"committer":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2019-02-26 08:36:37.000000000","tz":0},"subject":"Lease creation/updation should fail if amount is 0","message":"Lease creation/updation should fail if amount is 0\n\nLease is getting created and updated for resource_type\nvirtual:instance with amount\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with amount\u003d0, it will raise 400 BadRequest.\n\nLease creation for resource_type physical:host with\nmin\u003d0 or max\u003d0 raises 500 Internal Server Error\nand lease is updated for resource_type physical:host\nwith min\u003d0 or max\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with min\u003d0 or max\u003d0, it will raise 400 BadRequest.\n\nAPIImpact:\n1)Return 400 instead of 201 when lease is created with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n2)Return 400 instead of 500 when lease is created with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n3)Return 400 instead of 200 when lease is updated with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n4)Return 400 instead of 200 when lease is updated with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n\nCloses-Bug: #1807101\n\nChange-Id: I70e35ac7a524afb04077da5978c7563dec120639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e2f0f58665636c46ac06730bf0fd6789fce8d2c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e2f0f58665636c46ac06730bf0fd6789fce8d2c4"}]},"branch":"refs/heads/master"},"42a6cd27e986abc56cbaea81fd8ab5a81456c8a3":{"kind":"REWORK","_number":4,"created":"2019-03-19 09:35:54.000000000","uploader":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"ref":"refs/changes/26/630026/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/26/630026/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/26/630026/4"}}},"commit":{"parents":[{"commit":"000b10b23f1b49062fec98d070d9a055c0e7ed7e","subject":"Merge \"Move hosts back to freepool in case of failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/000b10b23f1b49062fec98d070d9a055c0e7ed7e"}]}],"author":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2018-12-21 14:00:18.000000000","tz":0},"committer":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2019-03-19 09:34:43.000000000","tz":0},"subject":"Lease creation/updation should fail if amount is 0","message":"Lease creation/updation should fail if amount is 0\n\nLease is getting created and updated for resource_type\nvirtual:instance with amount\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with amount\u003d0, it will raise 400 BadRequest.\n\nLease creation for resource_type physical:host with\nmin\u003d0 or max\u003d0 raises 500 Internal Server Error\nand lease is updated for resource_type physical:host\nwith min\u003d0 or max\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with min\u003d0 or max\u003d0, it will raise 400 BadRequest.\n\nAPIImpact:\n1)Return 400 instead of 201 when lease is created with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n2)Return 400 instead of 500 when lease is created with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n3)Return 400 instead of 200 when lease is updated with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n4)Return 400 instead of 200 when lease is updated with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n\nCloses-Bug: #1807101\n\nChange-Id: I70e35ac7a524afb04077da5978c7563dec120639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/42a6cd27e986abc56cbaea81fd8ab5a81456c8a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/42a6cd27e986abc56cbaea81fd8ab5a81456c8a3"}]},"branch":"refs/heads/master"},"fd5f0dc6577581a725e577b0207a90be4e6d77a8":{"kind":"REWORK","_number":5,"created":"2019-03-28 11:34:55.000000000","uploader":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"ref":"refs/changes/26/630026/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/26/630026/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/26/630026/5"}}},"commit":{"parents":[{"commit":"27c3aea4fe649bc0dd7f0caaf0384c35096619d3","subject":"Add a release note for the floating IP plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/27c3aea4fe649bc0dd7f0caaf0384c35096619d3"}]}],"author":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2018-12-21 14:00:18.000000000","tz":0},"committer":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2019-03-28 05:54:31.000000000","tz":0},"subject":"Lease creation/updation should fail if amount is 0","message":"Lease creation/updation should fail if amount is 0\n\nLease is getting created and updated for resource_type\nvirtual:instance with amount\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with amount\u003d0, it will raise 400 BadRequest.\n\nLease creation for resource_type physical:host with\nmin\u003d0 or max\u003d0 raises 500 Internal Server Error\nand lease is updated for resource_type physical:host\nwith min\u003d0 or max\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with min\u003d0 or max\u003d0, it will raise 400 BadRequest.\n\nAPIImpact:\n1)Return 400 instead of 201 when lease is created with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n2)Return 400 instead of 500 when lease is created with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n3)Return 400 instead of 200 when lease is updated with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n4)Return 400 instead of 200 when lease is updated with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n\nCloses-Bug: #1807101\n\nChange-Id: I70e35ac7a524afb04077da5978c7563dec120639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fd5f0dc6577581a725e577b0207a90be4e6d77a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fd5f0dc6577581a725e577b0207a90be4e6d77a8"}]},"branch":"refs/heads/master"},"26e91408f5652a2ef2382809926ad90893c2f38f":{"kind":"REWORK","_number":6,"created":"2019-04-03 08:29:38.000000000","uploader":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"ref":"refs/changes/26/630026/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/26/630026/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/26/630026/6"}}},"commit":{"parents":[{"commit":"c36dc9b652da23e67e68e7c424c0b2502307b08c","subject":"Merge \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c36dc9b652da23e67e68e7c424c0b2502307b08c"}]}],"author":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2018-12-21 14:00:18.000000000","tz":0},"committer":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2019-04-03 07:23:15.000000000","tz":0},"subject":"Lease creation/updation should fail if amount is 0","message":"Lease creation/updation should fail if amount is 0\n\nLease is getting created and updated for resource_type\nvirtual:instance with amount\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with amount\u003d0, it will raise 400 BadRequest.\n\nLease creation for resource_type physical:host with\nmin\u003d0 or max\u003d0 raises 500 Internal Server Error\nand lease is updated for resource_type physical:host\nwith min\u003d0 or max\u003d0.\nThis patch doesn\u0027t allow creation and updation of\nlease with min\u003d0 or max\u003d0, it will raise 400 BadRequest.\n\nAPIImpact:\n1)Return 400 instead of 201 when lease is created with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n2)Return 400 instead of 500 when lease is created with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n3)Return 400 instead of 200 when lease is updated with\namount\u003d0 for resource_type \u0027virtual:instance\u0027.\n4)Return 400 instead of 200 when lease is updated with\nmin\u003d0 or max\u003d0 for resource_type \u0027physical:host\u0027.\n\nCloses-Bug: #1807101\n\nChange-Id: I70e35ac7a524afb04077da5978c7563dec120639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/26e91408f5652a2ef2382809926ad90893c2f38f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/26e91408f5652a2ef2382809926ad90893c2f38f"}]},"branch":"refs/heads/master"},"fa6bb0b5b2976906214b7f7f4f4aa5cb24234736":{"kind":"REWORK","_number":7,"created":"2019-06-06 09:05:24.000000000","uploader":{"_account_id":29278,"name":"Asmita Singh","email":"Asmita.Singh@nttdata.com","username":"asmita.singh"},"ref":"refs/changes/26/630026/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/26/630026/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/26/630026/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/26/630026/7"}}},"commit":{"parents":[{"commit":"4e6727e3ebaa9b26c90db640ed61738bb3b9d3df","subject":"Merge \"Retry on inventory update conflict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4e6727e3ebaa9b26c90db640ed61738bb3b9d3df"}]}],"author":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2018-12-21 14:00:18.000000000","tz":0},"committer":{"name":"asmita singh","email":"asmita.singh@nttdata.com","date":"2019-06-06 07:35:35.000000000","tz":0},"subject":"Lease creation/update should fail if amount is 0","message":"Lease creation/update should fail if amount is 0\n\nLease is getting created and updated for resource_type virtual:instance\nwith amount\u003d0.\nThis patch doesn\u0027t allow creation and update of lease with amount\u003d0, it\nwill raise 400 BadRequest.\n\nLease creation for resource_type physical:host with min\u003d0 or max\u003d0\nraises 500 Internal Server Error and lease is updated for resource_type\nphysical:host with min\u003d0 or max\u003d0.\nThis patch doesn\u0027t allow creation and update of lease with min\u003d0 or\nmax\u003d0, it will raise 400 BadRequest.\n\nAPIImpact:\n1)Return 400 instead of 201 when lease is created with amount\u003d0 for\nresource_type \u0027virtual:instance\u0027.\n2)Return 400 instead of 500 when lease is created with min\u003d0 or max\u003d0\nfor resource_type \u0027physical:host\u0027.\n3)Return 400 instead of 200 when lease is updated with amount\u003d0 for\nresource_type \u0027virtual:instance\u0027.\n4)Return 400 instead of 200 when lease is updated with min\u003d0 or max\u003d0\nfor resource_type \u0027physical:host\u0027.\n\nCloses-Bug: #1807101\n\nChange-Id: I70e35ac7a524afb04077da5978c7563dec120639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fa6bb0b5b2976906214b7f7f4f4aa5cb24234736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/fa6bb0b5b2976906214b7f7f4f4aa5cb24234736"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
